)]}'
{"id":"openstack%2Fnetworking-bagpipe~667520","triplet_id":"openstack%2Fnetworking-bagpipe~master~I2187f0e63d366fe7d174d0f86bb548535e9eef49","project":"openstack/networking-bagpipe","branch":"master","topic":"bindep-fallback","hashtags":[],"change_id":"I2187f0e63d366fe7d174d0f86bb548535e9eef49","subject":"Add local bindep.txt","status":"MERGED","created":"2019-06-26 06:29:43.000000000","updated":"2019-06-28 20:28:05.000000000","submitted":"2019-06-28 20:28:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667520-1561753685046-971b6051","meta_rev_id":"339d771b70a4a892c7eadfb37d5fe0f1cd2283b5","_number":667520,"virtual_id_number":667520,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-06-28 20:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-06-28 18:37:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-06-27 21:33:40.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-06-28 18:37:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 21:33:40.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-06-28 18:37:24.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-06-28 20:28:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17ee2b1769f54617650b0d488797d9ae1be6ce30","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 06:29:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c72cd8b6b5fa7bf6e076d7c2b871ef5f099e9cc6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 06:32:11.000000000","message":"Patch Set 1:\n\nWithout this patch some tests fail (e.g.: [1]) with \u0027command not found\u0027 errors.\n\n[1] http://logs.openstack.org/22/636422/2/check/openstack-tox-py27/0d26f90/job-output.txt.gz#_2019-06-24_19_32_42_055816","accounts_in_message":[],"_revision_number":1},{"id":"6158d29e87248ffca097b3c2e8d15562c641e046","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 07:51:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b5ab0d1c3db44d7118b5431adbd90b9c088cbf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 09:49:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/20/667520/2/check/openstack-tox-pep8/94d46d9/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/20/667520/2/check/openstack-tox-py27/a996921/ : SUCCESS in 7m 18s\n- openstack-tox-py36 http://logs.openstack.org/20/667520/2/check/openstack-tox-py36/e5a0e29/ : FAILURE in 6m 26s\n- openstack-tox-py37 http://logs.openstack.org/20/667520/2/check/openstack-tox-py37/bb7862c/ : SUCCESS in 8m 12s\n- openstack-tox-docs http://logs.openstack.org/20/667520/2/check/openstack-tox-docs/13fc842/html/ : SUCCESS in 4m 27s\n- build-openstack-releasenotes http://logs.openstack.org/20/667520/2/check/build-openstack-releasenotes/5564429/html/ : SUCCESS in 3m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667520/2/check/openstack-tox-lower-constraints/6804c42/ : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/20/667520/2/check/openstack-tox-cover/c5659fe/cover/ : SUCCESS in 7m 32s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/20/667520/2/check/legacy-tempest-dsvm-networking-bagpipe/f4a16df/ : SUCCESS in 1h 03m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/20/667520/2/check/neutron-tempest-plugin-bgpvpn-bagpipe/15dbb91/ : SUCCESS in 1h 47m 54s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/20/667520/2/check/legacy-networking-bagpipe-dsvm-fullstack/1d62c38/ : FAILURE in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a31de46c035c61ed29f4173bd35e957e2f2504c1","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-27 18:56:15.000000000","message":"Patch Set 2:\n\nrecheck some odd recursion error with abc in py36 \u0027RecursionError: maximum recursion depth exceeded while calling a Python object\u0027","accounts_in_message":[],"_revision_number":2},{"id":"d5ab1b3962c5007967ba842fbe0fca14f7bc84fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 20:47:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/667520/2/check/openstack-tox-pep8/c7a1784/ : SUCCESS in 6m 09s\n- openstack-tox-py27 http://logs.openstack.org/20/667520/2/check/openstack-tox-py27/40b0ffb/ : SUCCESS in 6m 51s\n- openstack-tox-py36 http://logs.openstack.org/20/667520/2/check/openstack-tox-py36/c112cf0/ : SUCCESS in 8m 24s\n- openstack-tox-py37 http://logs.openstack.org/20/667520/2/check/openstack-tox-py37/20791bf/ : SUCCESS in 6m 58s\n- openstack-tox-docs http://logs.openstack.org/20/667520/2/check/openstack-tox-docs/42fae47/html/ : SUCCESS in 4m 14s\n- build-openstack-releasenotes http://logs.openstack.org/20/667520/2/check/build-openstack-releasenotes/ab6b93c/html/ : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/667520/2/check/openstack-tox-lower-constraints/ba0c54a/ : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/20/667520/2/check/openstack-tox-cover/be961f5/cover/ : SUCCESS in 6m 23s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/20/667520/2/check/legacy-tempest-dsvm-networking-bagpipe/8898c11/ : SUCCESS in 51m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/20/667520/2/check/neutron-tempest-plugin-bgpvpn-bagpipe/33ea78a/ : SUCCESS in 1h 42m 44s\n- legacy-networking-bagpipe-dsvm-fullstack http://logs.openstack.org/20/667520/2/check/legacy-networking-bagpipe-dsvm-fullstack/dfb88a9/ : FAILURE in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2224b1200d472428c4ac5bd33b39af80dba6cac5","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-27 21:33:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a5d589e9088706e78c7ec83c52f13ae8a2d53c47","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-28 18:37:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8ed6ef67ff9bcb66dbed440364f4cf302543fc62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 18:37:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b8087c369429c8d3fd42149a8c1c0f362b65a0cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 20:28:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/667520/2/gate/openstack-tox-pep8/3ae8652/ : SUCCESS in 6m 28s\n- openstack-tox-py27 http://logs.openstack.org/20/667520/2/gate/openstack-tox-py27/b52df54/ : SUCCESS in 6m 37s\n- openstack-tox-py36 http://logs.openstack.org/20/667520/2/gate/openstack-tox-py36/5f11bd5/ : SUCCESS in 6m 48s\n- openstack-tox-py37 http://logs.openstack.org/20/667520/2/gate/openstack-tox-py37/d1a6fb9/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/20/667520/2/gate/openstack-tox-docs/040d93a/html/ : SUCCESS in 4m 30s\n- build-openstack-releasenotes http://logs.openstack.org/20/667520/2/gate/build-openstack-releasenotes/e54b1d2/html/ : SUCCESS in 3m 31s\n- legacy-tempest-dsvm-networking-bagpipe http://logs.openstack.org/20/667520/2/gate/legacy-tempest-dsvm-networking-bagpipe/43c0f1e/ : SUCCESS in 59m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe http://logs.openstack.org/20/667520/2/gate/neutron-tempest-plugin-bgpvpn-bagpipe/793e24f/ : SUCCESS in 1h 50m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2557dd5db57aa0d15b25f922ce21786880970c9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 20:28:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f1a317df0aa09c8f66dfc2f8487f79e43ae33aa","revisions":{"73921be3162795b8e3014e89fc9a5e6d027cd450":{"kind":"REWORK","_number":1,"created":"2019-06-26 06:29:43.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/20/667520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/20/667520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/667520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/667520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/667520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/667520/1"}}},"commit":{"parents":[{"commit":"f1d1748e4b2d099690e5b550a80230d87cf30f24","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1d1748e4b2d099690e5b550a80230d87cf30f24"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2019-06-26 06:28:15.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2019-06-26 06:28:15.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I2187f0e63d366fe7d174d0f86bb548535e9eef49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/73921be3162795b8e3014e89fc9a5e6d027cd450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/73921be3162795b8e3014e89fc9a5e6d027cd450"}]},"branch":"refs/heads/master"},"3f1a317df0aa09c8f66dfc2f8487f79e43ae33aa":{"kind":"REWORK","_number":2,"created":"2019-06-26 07:51:41.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/20/667520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/20/667520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/667520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/667520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/667520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/667520/2"}}},"commit":{"parents":[{"commit":"f1d1748e4b2d099690e5b550a80230d87cf30f24","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1d1748e4b2d099690e5b550a80230d87cf30f24"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2019-06-26 06:28:15.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2019-06-26 07:51:01.000000000","tz":120},"subject":"Add local bindep.txt","message":"Add local bindep.txt\n\nAs it was announced [1] global bindep-fallback.txt was removed and now\nprojects need to have a local bindep.txt to be able to install binary\ndependencies for testing.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007272.html\n\nChange-Id: I2187f0e63d366fe7d174d0f86bb548535e9eef49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3f1a317df0aa09c8f66dfc2f8487f79e43ae33aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3f1a317df0aa09c8f66dfc2f8487f79e43ae33aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
