)]}'
{"id":"openstack%2Fnetworking-bagpipe~674158","triplet_id":"openstack%2Fnetworking-bagpipe~master~I528eded1f72d7d36aa6b99615fdb8bc1d5de3458","project":"openstack/networking-bagpipe","branch":"master","topic":"sphinx","hashtags":[],"change_id":"I528eded1f72d7d36aa6b99615fdb8bc1d5de3458","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","status":"MERGED","created":"2019-08-02 02:52:27.000000000","updated":"2019-08-26 23:22:56.000000000","submitted":"2019-08-26 23:21:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674158-1566861703226-1d226e13","meta_rev_id":"3106644fe643a538df8f92cfd8fc020c7a6284cc","_number":674158,"virtual_id_number":674158,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-08-26 23:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2019-08-13 17:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-08-26 21:39:15.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-08-26 21:39:15.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 17:49:31.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-08-26 21:39:15.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-08-26 23:21:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6205da40e74432bdc487a19ae59a434afd17686","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-02 02:52:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"202bcbeff8bfd77dcfc88a0c892bff418d99f75d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 04:30:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/58/674158/1/check/openstack-tox-pep8/3533399/ : SUCCESS in 7m 13s\n- openstack-tox-docs https://logs.opendev.org/58/674158/1/check/openstack-tox-docs/d640472/html/ : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://logs.opendev.org/58/674158/1/check/build-openstack-releasenotes/6b986c9/html/ : SUCCESS in 3m 06s\n- requirements-check https://logs.opendev.org/58/674158/1/check/requirements-check/505ee2f/ : SUCCESS in 3m 05s\n- legacy-tempest-dsvm-networking-bagpipe https://logs.opendev.org/58/674158/1/check/legacy-tempest-dsvm-networking-bagpipe/75a8b2d/ : SUCCESS in 55m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://logs.opendev.org/58/674158/1/check/neutron-tempest-plugin-bgpvpn-bagpipe/eb4693e/ : SUCCESS in 1h 36m 24s\n- legacy-networking-bagpipe-dsvm-fullstack https://logs.opendev.org/58/674158/1/check/legacy-networking-bagpipe-dsvm-fullstack/9356375/ : FAILURE in 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37c1d837a444ca1c207dc98abac41b3ebf6c2a63","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-13 17:49:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c9ee25fb00f90fde8944230dec3c4ae35cdd4b0","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-08-26 21:39:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dca1efda54bc7ca7786ca5dfce00c8bc40f8ecc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 21:39:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"657357c7f018ef32f4df08ec97dffadc2ee3721e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 23:21:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e00b62913c94ce2b81da764fa9000a0 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d1f1705d2f4e4991fc50f49dcbc8c2 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57f378f3cfde40afb3113d2e9485d9a7 : SUCCESS in 5m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/71813253a3904e3c94be90aedb4a2104 : SUCCESS in 4m 21s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7446d2e3fa3742fd85ed56932b2bd1bd : SUCCESS in 1h 39m 26s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/353d0cf895d04764bf8043c31b2d4f29 : SUCCESS in 1h 29m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f0109a485665208fbf2cb974fb14231aacb9d4ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 23:21:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4093baa0c343a2a1b49cab8f94cdbbfb007ec5c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 23:22:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92321d22ef054fdf992de4698b10e855 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a5d93ac5823bfc59766d5d12d88bdf1661a18f7","revisions":{"9a5d93ac5823bfc59766d5d12d88bdf1661a18f7":{"kind":"REWORK","_number":1,"created":"2019-08-02 02:52:27.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/58/674158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/58/674158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/58/674158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/58/674158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/58/674158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/58/674158/1"}}},"commit":{"parents":[{"commit":"f8139b0fbb7fedd787969e18c67c65bcd6869402","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f8139b0fbb7fedd787969e18c67c65bcd6869402"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 02:51:39.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 02:52:43.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\n\nChange-Id: I528eded1f72d7d36aa6b99615fdb8bc1d5de3458\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9a5d93ac5823bfc59766d5d12d88bdf1661a18f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9a5d93ac5823bfc59766d5d12d88bdf1661a18f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
