)]}'
{"id":"openstack%2Fnetworking-bagpipe~677648","triplet_id":"openstack%2Fnetworking-bagpipe~master~I7c094940c6bde13ec91ac0f36bc9303ce46bad29","project":"openstack/networking-bagpipe","branch":"master","topic":"python3-first","hashtags":[],"change_id":"I7c094940c6bde13ec91ac0f36bc9303ce46bad29","subject":"Ensure to python3 as basepython in all tox envs","status":"MERGED","created":"2019-08-21 06:42:35.000000000","updated":"2019-08-27 16:31:56.000000000","submitted":"2019-08-27 16:28:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677648-1566923302840-93e58965","meta_rev_id":"6dab0bcc91b4fd50afd05f479ec3358674c9e302","_number":677648,"virtual_id_number":677648,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-08-27 16:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2019-08-23 09:13:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-08-27 14:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-08-27 14:19:46.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-08-27 14:24:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 09:13:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-08-27 14:19:46.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-08-27 14:24:42.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-08-27 16:28:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cdde4c6112d8b3623bcd7adc18829c7999d9aa15","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-21 06:42:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88946549fa6602bc3aff190080228bf4d19a029f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 08:21:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b936e473b3524e25b5e4d682bbee3f63 : SUCCESS in 8m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/591674d262734314ae1c79a4815192eb : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d65f42679eec4233871153fa8fb388fd : SUCCESS in 9m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c312b9c0d8d46c89e499b4e6efb32b0 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70b570d6e78c49d79c3c5703add3e8a2 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/713a3cab3a7442cab32b5ac63a19ef05 : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4043ee7017bc456a999ef369b5ef6f26 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/753ea38c413744458cc15dc06cded46b : SUCCESS in 7m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3c7f8e2b94a841e38a11706936e816ea : SUCCESS in 1h 35m 51s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/96eb38fb204e4e96b622b6d4f93eb589 : FAILURE in 15m 14s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/bbbb9320c287414abda8eece8b0be540 : SUCCESS in 1h 26m 27s","accounts_in_message":[],"_revision_number":1},{"id":"93840bde6bd9b16a80b2f3a23a8057e9d2358c00","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-23 09:13:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"847f5f69d055ea0bc61257dc50bb4d82990108ae","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-08-27 14:19:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9d0dbbada1c138592c3b2510a5e4ded6ef71350d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-08-27 14:24:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7de1b34482fcd039999fccfba38d61ec646de0b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 14:24:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abed005d7aa5e94ded3479cfe95e23b50dc43533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 16:28:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6a181e6b1740b589faeff4c922d1f8 : SUCCESS in 9m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d3c5c46b33f4194948c0a2e6abf678f : SUCCESS in 9m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b707f004c8294f1ab3e1517d0b49081b : SUCCESS in 9m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a25e0e61c0e46bbbbaa052670abdb6d : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d2065c89324f60833ad5140ce461b9 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ee751932bff417ca32c62d2262203f6 : SUCCESS in 8m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/aad9e28322c846dfb21d35a545870cb3 : SUCCESS in 1h 48m 22s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b351209f98b747b7b6fbf62fcb7a85e2 : SUCCESS in 1h 40m 21s","accounts_in_message":[],"_revision_number":1},{"id":"247a3d7f1f1c0c47e8938c5ebf30b2883dc363e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 16:28:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ea4c0d7cb74cf400c5a96b77e2be8c8b12b0b3c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 16:31:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/066dc109ad464c61bfafbad2353ad5db : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ee826f36adbc4aa4c5d8c73d6fb6199fbc843d8","revisions":{"2ee826f36adbc4aa4c5d8c73d6fb6199fbc843d8":{"kind":"REWORK","_number":1,"created":"2019-08-21 06:42:35.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/48/677648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/677648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/677648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/677648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/677648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/677648/1"}}},"commit":{"parents":[{"commit":"6efee739435090e8bf1412d1ea0f5b66a5abcfa3","subject":"Merge \"Upgrade pylint to a version that works with python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6efee739435090e8bf1412d1ea0f5b66a5abcfa3"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-21 06:36:10.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-21 06:40:24.000000000","tz":540},"subject":"Ensure to python3 as basepython in all tox envs","message":"Ensure to python3 as basepython in all tox envs\n\nThere are some remaining tox envs with python2.7.\nThis commit ensures to specify python3 as basepython\nfor all meaningful tox environments.\n\nNote that fullstack job is broken now due to unrelated reason(s)\nand it will be fixed separately, so do not mind the test result.\n\nAnother note is the functional job is not run in the gate.\nIt seems because the functional job has only migration tests\nand it is covered by \u0027check_migration\u0027 in the pep8 job.\nAnyway it is nice to run the functional job with python3.\n\nChange-Id: I7c094940c6bde13ec91ac0f36bc9303ce46bad29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2ee826f36adbc4aa4c5d8c73d6fb6199fbc843d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2ee826f36adbc4aa4c5d8c73d6fb6199fbc843d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
