)]}'
{"id":"openstack%2Fnetworking-bagpipe~683337","triplet_id":"openstack%2Fnetworking-bagpipe~master~I42af2e0055e53d07be24224c5d779035ee92d1f7","project":"openstack/networking-bagpipe","branch":"master","topic":"I42af2e0055e53d07be24224c5d779035ee92d1f7","hashtags":[],"change_id":"I42af2e0055e53d07be24224c5d779035ee92d1f7","subject":"Update the constraints url","status":"MERGED","created":"2019-09-20 09:31:47.000000000","updated":"2019-10-10 15:22:16.000000000","submitted":"2019-10-10 15:20:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683337-1570720855655-fe29816b","meta_rev_id":"2035e1277a3c7b6a7f6a5fb1359ed178857427e3","_number":683337,"virtual_id_number":683337,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 15:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2019-10-09 06:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2019-10-10 13:45:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2019-10-10 13:45:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 06:34:15.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-10-10 13:45:26.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-10-10 15:20:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"684c09769e367abe34cf902136eba9e010ed3267","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-20 09:31:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2fd9539daefd59daab302b3e7e5db3c40a232e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 11:27:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c071957440404bffbff8f175f8df3cc3 : SUCCESS in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dee978cefd274aa29031b350005ca8f5 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d49d008ef6c64e41b36a3729e00a1cca : SUCCESS in 7m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40633e44d54343d38a93a6330870f9bf : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c56038b3bd2246708b69ab70e248a572 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2a45eed200849fd864dbaa058da5178 : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/818240dce4c74124be7ac3c6d1919710 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d53ff5147f74356a56ede9a14b333f1 : SUCCESS in 6m 38s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/03f3a5cfb2cd4ae29f37888b5717a783 : SUCCESS in 1h 52m 24s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/91b4141972b2422d83c581780018e327 : FAILURE in 26m 45s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/bbf68ecab763416a9d76fca91924276b : SUCCESS in 1h 29m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b75e20cd5a1ace9cfe01ada34f9966203a8ab1d0","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-03 19:36:51.000000000","message":"Patch Set 1: Code-Review-1\n\nA reference also needs to be updated here: http://codesearch.openstack.org/?q\u003dupper-\u0026i\u003dnope\u0026files\u003d%5Edoc\u0026repos\u003dopenstack/networking-bagpipe","accounts_in_message":[],"_revision_number":1},{"id":"9983f83aaf3153bcf412d1ba24ba2cbb531b9593","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-08 07:07:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48eb418010b926110256f0729433e2daebe5a79e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 09:17:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af6ab7ffcfd042eebd0e8a9e7b413a6e : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de3b38c6a55e43719ad178b3da47b40f : SUCCESS in 12m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bdf1f5cd596421db345c061cedbba25 : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7f9458d812c4fc8af6a08bad0d7a41c : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c0e34fbaea41bcabfbf20f5db24671 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd549ab0f795460fac96ba4509bf51e9 : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cb61ca4055b4bc589c0a6c6ec2f6f64 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6825ca712b08494dbdbb8dbf20109c2c : SUCCESS in 10m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/388d2b60149f4c59b3515b88cb39c346 : SUCCESS in 2h 06m 22s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e1ed2e6aafcc4a6b877652258e53db71 : FAILURE in 24m 16s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/8fee51226da24afd921d7fcada14db68 : SUCCESS in 1h 32m 58s","accounts_in_message":[],"_revision_number":2},{"id":"18caa8343dfba972a47b4f63f95caeb96ad48c83","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-08 19:14:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d4669b8337079f09ec90d56694273f3cd6e6421","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-09 01:36:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cbbd8e0ef1b8adccedebd2e8fea431c6638925b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 02:55:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a94dd70d678496ab8ac30d131565083 : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56ec68cee9fb4ee5be970860fb802eea : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d134bbbcc9fd4fc0bd65f8e529a5bf9b : SUCCESS in 6m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc45f2eea04948b08732f7340998aaa3 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9748284e3a6d4cf7a4f63aecd2912a52 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c3b27c3828f442899ecb262295fa3a7 : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85457633a70d410f8717cd462e247ddc : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b061243ee30a454cb5e3440a84c391e8 : SUCCESS in 7m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/758ab8a988e0430f8dc8c0b983791b58 : SUCCESS in 1h 16m 02s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/951ac0fbd09046f88b30127a3556f620 : FAILURE in 19m 23s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/cce90f5c755a4357a90e00dbc3fca40e : SUCCESS in 1h 10m 27s","accounts_in_message":[],"_revision_number":3},{"id":"a0c2d9bf0a5fcad166f677f0be8c06dc88d293f6","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-09 06:34:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d899762e9b8ab802097d2ec17aed4f89f23d6d68","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-10 13:45:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7cc92734e4fb2ba320a0c543f384a2f0ec6e20e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:45:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d2be73352c3e0aa3c8b84a2db30e9d01a517a2df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:20:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c1faca044523c7e5232cd56467575cf22caef3c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:20:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ae235aa43b456cbdb030231e78368d : SUCCESS in 6m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6156bb9b084d437292850ed5e40920ce : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f0289f74ec24024a1d9cd82776d16f6 : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6ec94f2c79a46ecbc24d9e71da575f8 : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c4dd6dd91c4ce7b3e26dcc78ecc0b9 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c6364f136f6401db14d72bd72af68fe : SUCCESS in 4m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6a0c54144d0c450db98dc64bd8877521 : SUCCESS in 1h 32m 10s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/813f59ed66164d3d8ec58c122b25b00d : SUCCESS in 1h 26m 04s","accounts_in_message":[],"_revision_number":3},{"id":"60b5415787e51206fd5c329558627ab1b78d7224","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:22:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4990cca57f4498a9ba2ca66512d0110 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb909b02f0d7402584bde442dc11b3e9 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cc6485b951fbad7d48bd8edcbcacbfd972433473","revisions":{"2c86bce11dc020d68e37bfa8c10e5f1ff59873ac":{"kind":"REWORK","_number":1,"created":"2019-09-20 09:31:47.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/37/683337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/37/683337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/1"}}},"commit":{"parents":[{"commit":"f9a31f5c01f93f9675c99917f39e7a72956de69c","subject":"Merge \"Ensure to python3 as basepython in all tox envs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f9a31f5c01f93f9675c99917f39e7a72956de69c"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-20 09:32:06.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-20 09:32:09.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I42af2e0055e53d07be24224c5d779035ee92d1f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2c86bce11dc020d68e37bfa8c10e5f1ff59873ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2c86bce11dc020d68e37bfa8c10e5f1ff59873ac"}]},"branch":"refs/heads/master"},"6df21971d6c3965d4233302047b3634e0ea2932f":{"kind":"REWORK","_number":2,"created":"2019-10-08 07:07:33.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/37/683337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/37/683337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/2"}}},"commit":{"parents":[{"commit":"f9a31f5c01f93f9675c99917f39e7a72956de69c","subject":"Merge \"Ensure to python3 as basepython in all tox envs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f9a31f5c01f93f9675c99917f39e7a72956de69c"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-20 09:32:06.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-08 07:07:34.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I42af2e0055e53d07be24224c5d779035ee92d1f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6df21971d6c3965d4233302047b3634e0ea2932f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6df21971d6c3965d4233302047b3634e0ea2932f"}]},"branch":"refs/heads/master"},"cc6485b951fbad7d48bd8edcbcacbfd972433473":{"kind":"REWORK","_number":3,"created":"2019-10-09 01:36:16.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/37/683337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/37/683337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/683337/3"}}},"commit":{"parents":[{"commit":"f9a31f5c01f93f9675c99917f39e7a72956de69c","subject":"Merge \"Ensure to python3 as basepython in all tox envs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f9a31f5c01f93f9675c99917f39e7a72956de69c"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-20 09:32:06.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-09 01:36:39.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I42af2e0055e53d07be24224c5d779035ee92d1f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/cc6485b951fbad7d48bd8edcbcacbfd972433473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/cc6485b951fbad7d48bd8edcbcacbfd972433473"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
