)]}'
{"id":"openstack%2Fnetworking-bagpipe~685230","triplet_id":"openstack%2Fnetworking-bagpipe~master~Iba6c78d7e1de9fb175a6c4bb47f5ac62ce05cde4","project":"openstack/networking-bagpipe","branch":"master","topic":"train-release","hashtags":[],"change_id":"Iba6c78d7e1de9fb175a6c4bb47f5ac62ce05cde4","subject":"Switch neutron-tempest-plugin jobs to train branch","status":"ABANDONED","created":"2019-09-27 08:23:43.000000000","updated":"2019-10-03 20:07:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7e8422a0c270c85c978516cdc48dda74b119e6a5","_number":685230,"virtual_id_number":685230,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-03 19:38:05.000000000","_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"date":"2019-10-03 20:06:47.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-09-30 14:50:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-30 14:50:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-03 19:38:05.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"}],"messages":[{"id":"39c8cca1bb4ec3597e692c4712e41cc13d4e2101","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-27 08:23:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"924f82292cbfc394bf32be8861cb0c1a9be18071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 08:28:31.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"a7d880768b1cce0ba7ed4137305c7392b9a9fd80","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-27 08:29:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5f5db7bc62ce4a6e7f37060f314177f2a0120895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 10:17:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d1ed9f232e84f5aa374f8db2700e873 : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2ca7fd8657342868775defd33f57fa7 : FAILURE in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7fa31bd20ec468db1720fb567fac432 : SUCCESS in 8m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29421d49f19a459695d1b64a22af89e4 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf7e9dc3198e49beb69fb504feb2be33 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6aad94904af42fdb25008ded3e582af : FAILURE in 6m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a882307b92dc42b5abe735ace799dcbb : SUCCESS in 7m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/514401e9a0e3474ab30b7bb53b543369 : SUCCESS in 1h 32m 22s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ac90b1f1ed164129aa0f4bdc48b8d776 : FAILURE in 25m 03s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/9b342a0a0f9347dea784a2b56ae65b17 : SUCCESS in 1h 32m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c29ccbd80c3ac90c78b79a2df767d5fb91b2ad31","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-30 13:05:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"784e0ba5c89426eab3a4852f91673322f1ebc7be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 14:50:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d798020d28784c7081755a538d8abc59 : SUCCESS in 7m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce1ab1e42a6e4325b1a4cfb75596e799 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24ff7929817a4c7d8cfa0a43320a9c41 : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/601fa547384944249747816dfa8034fe : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7c7d8ddb764db19f9850e878cb4750 : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff5ff355bf2c4d779f7597ea0423505e : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6016506f2f8f42d5a5606b6c9a2d7470 : SUCCESS in 7m 31s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/28cfd96aec0d45d0b63e92ba275f14fe : SUCCESS in 1h 39m 14s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ac576ec9c1dd4c7c873e991a37fff6d8 : FAILURE in 20m 19s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/221127316658484ab9b6969395a6ed51 : SUCCESS in 1h 27m 22s","accounts_in_message":[],"_revision_number":1},{"id":"52c5c26858b99b766f25991fd29624a0df00c85e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-02 11:21:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a62bfcef387f8ea3b97b1fd2d839ea2316a7b864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 13:52:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73636a89da14c849ae0209e43bbaafe : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c96dfb467d7140cd9018d1e6b2b167b9 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/489279059bf4477989f23e282761ec09 : SUCCESS in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f08f75e0e6ea4646a8d124b3c23962b9 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b55422989640368af468fc30763e4b : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6a826909f074aa6afbc8601d6576e37 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87c61fdb01a94ccdbc4afb4d248972cc : SUCCESS in 9m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/e9ee503d6ab24773a7a35eceb2b3e4a3 : SUCCESS in 1h 48m 05s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/406c4ed59bdd42ed88f1dd67aaecab78 : FAILURE in 22m 55s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/22a9a3dee0ff420f8d3a78659d97f7c6 : SUCCESS in 1h 17m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d1add0aae7f79ca0f08c59fff9457448a2305cc5","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-03 19:38:05.000000000","message":"Patch Set 1:\n\nShouldn\u0027t this change target stable/train branch ?","accounts_in_message":[],"_revision_number":1},{"id":"90030ea731969548ab24225a48eeb653c045b656","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-03 20:06:47.000000000","message":"Patch Set 1:\n\n\u003e Shouldn\u0027t this change target stable/train branch ?\n\nYou\u0027re totally right. I was probably tired or not woken up enough yet when I sent those patches :) Patch for train is in https://review.opendev.org/686483","accounts_in_message":[],"_revision_number":1},{"id":"fe8f89a1b33ca6d3429646eadf117ca3b082b8cc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-03 20:07:00.000000000","message":"Abandoned\n\nIt\u0027s needed for train branch instead of master","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4517a6636cb75ba7d10581e3235f065d920c686","revisions":{"f4517a6636cb75ba7d10581e3235f065d920c686":{"kind":"REWORK","_number":1,"created":"2019-09-27 08:23:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/30/685230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/30/685230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/685230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/685230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/685230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/685230/1"}}},"commit":{"parents":[{"commit":"f9a31f5c01f93f9675c99917f39e7a72956de69c","subject":"Merge \"Ensure to python3 as basepython in all tox envs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f9a31f5c01f93f9675c99917f39e7a72956de69c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-27 08:22:52.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-27 08:23:41.000000000","tz":120},"subject":"Switch neutron-tempest-plugin jobs to train branch","message":"Switch neutron-tempest-plugin jobs to train branch\n\nDepends-On: https://review.opendev.org/#/c/685213/\nChange-Id: Iba6c78d7e1de9fb175a6c4bb47f5ac62ce05cde4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f4517a6636cb75ba7d10581e3235f065d920c686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f4517a6636cb75ba7d10581e3235f065d920c686"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
