)]}'
{"id":"openstack%2Fnetworking-bagpipe~685598","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Frocky~I4589871c7482c8c317c323d9e886169ab8018bd2","project":"openstack/networking-bagpipe","branch":"stable/rocky","hashtags":[],"change_id":"I4589871c7482c8c317c323d9e886169ab8018bd2","subject":"Update Zuul releasenotes setup","status":"MERGED","created":"2019-09-29 15:29:50.000000000","updated":"2019-09-30 09:47:22.000000000","submitted":"2019-09-30 09:47:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685598-1569836842190-fe29816b","meta_rev_id":"902d654f2028773a5f309e9e0e2c2df8ddcd38f5","_number":685598,"virtual_id_number":685598,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-30 09:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2019-09-30 08:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-09-30 06:57:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"date":"2019-09-29 19:00:04.000000000","_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2019-09-30 08:25:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-29 19:00:04.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-09-30 06:57:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-09-30 08:25:29.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-09-30 09:47:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18e2c40631a20fe0064d77daabdd6525a7c69b54","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 15:29:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9b68e1499d896aded061409fe4c78b90ac83973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 17:01:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b85e0ff5fbc4e22a63f882d8730e1f2 : SUCCESS in 6m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e032dc4119fa46299ea6b0d957d389b0 : SUCCESS in 8m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8dcb95581f4f403598407367325cdc71 : SUCCESS in 6m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0b2e66d175264a3fa93a55d5a36c5ca5 : SUCCESS in 8m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b261156472c4d95bd1b547bb803ddc8 : SUCCESS in 7m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9672540384ce48069f491667f278eb54 : SUCCESS in 7m 17s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/168066ea5c3b456f8ab8e85d3d3bb927 : SUCCESS in 52m 37s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2a94a574a2a84b2589717259d353d738 : SUCCESS in 1h 22m 48s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/aa39e7c9a10743039184d520f3583786 : FAILURE in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a6dda32db278dbca9af2ddb048c053693e03fd6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-30 06:57:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"89f24a7bd8bd8aee39eea4a0b28729d48b803f02","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-30 08:25:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ac5d29892688392975bb790b60016113f06f790","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 08:25:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbce03798b0bd0023464a2dbe73176be0911d299","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 09:47:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb7a00bda94f4ab094fe47522526ffe1 : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9db909fe397847f8bb251fa80baa36fc : SUCCESS in 7m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b3451e4fa5a940a68efe84e012ff19e5 : SUCCESS in 6m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b142d9e6f30f4e1baa758aa97e729153 : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53e9a6afdf5a451f9b620da4198ef505 : SUCCESS in 7m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05c4395bcfd9411da0149f4f9b8f5e76 : SUCCESS in 8m 32s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/86f52dbafd08424f8f7df0ec7f6f7300 : SUCCESS in 58m 32s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/519303512e16457faf68ec164baeada8 : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"902d654f2028773a5f309e9e0e2c2df8ddcd38f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 09:47:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dfe50a56c2c882254e332ce62220051725a970b1","revisions":{"dfe50a56c2c882254e332ce62220051725a970b1":{"kind":"REWORK","_number":1,"created":"2019-09-29 15:29:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/98/685598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/685598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/685598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/685598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/685598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/685598/1"}}},"commit":{"parents":[{"commit":"a6037ee5e4acda4e330d0b92df5fabc75540ec68","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a6037ee5e4acda4e330d0b92df5fabc75540ec68"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:27:59.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:27:59.000000000","tz":120},"subject":"Update Zuul releasenotes setup","message":"Update Zuul releasenotes setup\n\nThe release-notes-jobs template is setup already, remove\nindvidual jobs - they are redundant.\n\nThis is a partial backport of I5fbf32b52ec6a5878761e23a5aee592e8c516bca,\nit is needed to cleanup the Zuul jobs.\n\nChange-Id: I4589871c7482c8c317c323d9e886169ab8018bd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dfe50a56c2c882254e332ce62220051725a970b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dfe50a56c2c882254e332ce62220051725a970b1"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
