)]}'
{"id":"openstack%2Fnetworking-bagpipe~685600","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fpike~I4589871c7482c8c317c323d9e886169ab8018bd2","project":"openstack/networking-bagpipe","branch":"stable/pike","topic":"pike/backport/685598","hashtags":[],"change_id":"I4589871c7482c8c317c323d9e886169ab8018bd2","subject":"Update Zuul releasenotes setup","status":"MERGED","created":"2019-09-29 15:32:32.000000000","updated":"2019-09-30 10:13:40.000000000","submitted":"2019-09-30 10:13:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685600-1569838420927-654589fa","meta_rev_id":"bc837c92b48d0d8a07c58024e7ea7caafe172115","_number":685600,"virtual_id_number":685600,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-30 10:13:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2019-09-30 09:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-09-30 06:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"date":"2019-09-29 19:00:16.000000000","_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2019-09-30 09:12:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-29 19:00:16.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-09-30 06:57:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-09-30 09:12:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-09-30 10:13:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23088c85582666cdb2aa1a0820d2626b221dbbc4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 15:32:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a71ceb00396fa8245f7e2e255808cd52aff1e595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 16:41:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed2a9324d0b9419fa563509cf9c508b5 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bcb2c35ee1eb465aaeb93aea1f58b2da : RETRY_LIMIT in 3m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/49e17a717b8f4064b12d7a18c6f31935 : RETRY_LIMIT in 2m 25s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/33bb16f4f77c4277a7401f36436e8ce7 : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1f9d09520ec41f48243a921a9f0a301 : RETRY_LIMIT in 1m 54s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/abd773528f04482e9628b58934135016 : SUCCESS in 1h 08m 33s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1ef63c38d25b4d6db6bb3dbfd2fe7358 : SUCCESS in 26m 12s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/c58e6bb4094b499890028a674b80c0e0 : FAILURE in 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f7ce0f05ee2b4290dbda0513e1ee9239cb6405a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-29 17:38:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ce4ca3fc31990ccc327ab173c2ccc4fb5f7eac99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 18:44:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f35b8fc3174f56b8aa3767cd9d4676 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef7cceb6978b4ce882d001b4d00f98e9 : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3879091697d14136beebdaf186f1b1fc : SUCCESS in 5m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b5b58143f5ae48ccbf3b957544456fff : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46885aa38e424dc5a4e6507b3d3f6648 : SUCCESS in 5m 11s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/c38abce74e0a45c6b5d5ed3cf788d90d : SUCCESS in 1h 00m 52s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/89c723a28e974e2a9e445bd23f8e48e7 : SUCCESS in 27m 39s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ef7de9661f604896b57e8062e64fbbb4 : FAILURE in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dd8a60e9537e39d1d303809af43d43dfe1ef2106","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-30 06:57:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1645ddc800e59e7b25c6e13c5487be24fb6230ed","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-30 09:12:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d8eb3e75ea7693273dcb69b0bbfee87e8c06aead","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 09:12:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d535b8f1e4db352b2d010c726b0069bd82b541bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 10:13:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531d36a5fc89434ab33e57a71548e724 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b955f9537134cdc9bd6184cd5351f50 : SUCCESS in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/67846ad81bcc498a89f9849a12523f67 : SUCCESS in 5m 38s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e4a06e55e90e499aad8ba70e281c392e : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dda00b606f04c79bcc4d1b77002ce0a : SUCCESS in 5m 29s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/b21d1d4931e44594a3f43610bcbcb7e8 : SUCCESS in 58m 39s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a24738addbc64be5b8664128c4041a3c : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0e3be44a8a1eb41a0828d2870aba32dd2bb8c033","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 10:13:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30eada2fb5b9bbe3d66dc6a98e1b66cf155bb077","revisions":{"71c6e0e10eef7289943f32b87a2e599fd7d39fd3":{"kind":"REWORK","_number":1,"created":"2019-09-29 15:32:32.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/00/685600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/685600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/685600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/685600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/685600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/685600/1"}}},"commit":{"parents":[{"commit":"e14e0acb4fa3f6d2be6d23939e4d5ed47d2f2773","subject":"Remove openstack-tox-py35-with-neutron-lib-master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e14e0acb4fa3f6d2be6d23939e4d5ed47d2f2773"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:27:59.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:32:18.000000000","tz":120},"subject":"Update Zuul releasenotes setup","message":"Update Zuul releasenotes setup\n\nThe release-notes-jobs template is setup already, remove\nindvidual jobs - they are redundant.\n\nThis is a partial backport of I5fbf32b52ec6a5878761e23a5aee592e8c516bca,\nit is needed to cleanup the Zuul jobs.\n\nChange-Id: I4589871c7482c8c317c323d9e886169ab8018bd2\n(cherry picked from commit dfe50a56c2c882254e332ce62220051725a970b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/71c6e0e10eef7289943f32b87a2e599fd7d39fd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/71c6e0e10eef7289943f32b87a2e599fd7d39fd3"}]},"branch":"refs/heads/stable/pike"},"30eada2fb5b9bbe3d66dc6a98e1b66cf155bb077":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-29 17:38:08.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/00/685600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/685600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/685600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/685600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/685600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/685600/2"}}},"commit":{"parents":[{"commit":"f59efc4297e144bbdf59b5d019e4936eaa156432","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f59efc4297e144bbdf59b5d019e4936eaa156432"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-29 15:27:59.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-09-29 17:38:08.000000000","tz":0},"subject":"Update Zuul releasenotes setup","message":"Update Zuul releasenotes setup\n\nThe release-notes-jobs template is setup already, remove\nindvidual jobs - they are redundant.\n\nThis is a partial backport of I5fbf32b52ec6a5878761e23a5aee592e8c516bca,\nit is needed to cleanup the Zuul jobs.\n\nChange-Id: I4589871c7482c8c317c323d9e886169ab8018bd2\n(cherry picked from commit dfe50a56c2c882254e332ce62220051725a970b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/30eada2fb5b9bbe3d66dc6a98e1b66cf155bb077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/30eada2fb5b9bbe3d66dc6a98e1b66cf155bb077"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
