)]}'
{"id":"openstack%2Fnetworking-bagpipe~688585","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fstein~Ic34aff147faac500c792146fa945f639c9c8992c","project":"openstack/networking-bagpipe","branch":"stable/stein","topic":"bug/1847605-stable/train-stable/stein","hashtags":[],"change_id":"Ic34aff147faac500c792146fa945f639c9c8992c","subject":"bagpipe-bgp: fix a broken looking-glass path","status":"MERGED","created":"2019-10-15 00:16:55.000000000","updated":"2019-10-24 19:53:03.000000000","submitted":"2019-10-24 19:51:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688585-1571946717661-fa70c53b","meta_rev_id":"f200240856a5bc3484d6eb3e0db096313fb2008e","_number":688585,"virtual_id_number":688585,"owner":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-24 15:25:17.000000000","_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 19:51:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-10-24 12:24:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-10-23 02:18:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"date":"2019-10-22 15:47:50.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2019-10-24 12:24:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 15:47:50.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-10-23 02:18:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-24 12:24:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-10-24 19:51:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fd9ee3c7a5ed90386f172921a11df23ee8108f2","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-15 00:16:55.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"de0d6f26a6693c513ee7a67d19d12d4ab9059c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 01:52:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7525dfb7215f453b9a664bfb035b53bf : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc95d2fa02e14f3bb199d32c5d75b2a7 : SUCCESS in 6m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/30fb36e67cb046d1aa5afd831b2c36b0 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37e5f7b8b1a2469383179e41c1ebf866 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7194aac92394a0993630628415e9c9f : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/539153072ce4413981f0d7edd2e46708 : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca5084a6b14741d0bc6ff953fa51691e : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b33514c80283426c80dddb2a23574199 : SUCCESS in 8m 09s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/299ea95331734aada0927d6f4052a89c : SUCCESS in 55m 21s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/fdf5e78e81944842bd536b72789d52fc : SUCCESS in 1h 30m 54s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/9246285c372f4da8b3956b245980a9e1 : FAILURE in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f34995e6f2812a0a7aa50235e17fbcab98343968","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-23 02:18:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"68a14ba8f037a634ee4af2e497204ddae38a1ea0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-24 12:24:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"723754a97a785383950b7a1860d1cbf8eeea7488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 12:24:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06f90b8b89b2f1e0fd095c5ec2783dcfa09a2a1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 14:03:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe13fb4215c24d3ebc46d7b73e4c2350 : FAILURE in 7m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e310a1d82aa14ab784539623f35de24e : FAILURE in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/938c6d3d25c14a8589e5828541442a23 : FAILURE in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab9c5f245bcf463cb12630053bb2bc4a : FAILURE in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ef4156284c4492ea67232da93e42c86 : FAILURE in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83833f7fc6284f6fb35a46b88f033d0e : SUCCESS in 4m 59s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/746e642a012346eca2e8590c038ba03b : SUCCESS in 54m 49s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/36102d56b6314ea59dd12d34bb4ed321 : SUCCESS in 1h 23m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6d1d03b3b21951346e2afc7b2443f34534494beb","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-24 15:25:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4a83e1c3bec9b7be80aca22d62585b0b4b7cd55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 18:18:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cba7685654f4180936cce64f73c8a4b : SUCCESS in 8m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fb50d19c20e41ba8f9b0fd35400ec68 : SUCCESS in 8m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93e071c5c1a84952b3d6c1be569efa02 : SUCCESS in 11m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f72afa09f664827a7b07ed1df183f72 : SUCCESS in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c57d093086174561b48c08fdf0393123 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c9d6cf0b9b454b9f0ec87fb4a2173f : SUCCESS in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aca7e5a56d054dd085517fc3943c4899 : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e2f35eb06534ad38443b6050e88083b : SUCCESS in 7m 52s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/74e4731fff884f028936835035c9d027 : SUCCESS in 52m 37s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4a3da48f7fd443b48ce73dc84154e697 : SUCCESS in 1h 21m 10s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f0e70e07c517422399ff6256399faaad : FAILURE in 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25c7467047c2d6f6af69ead6f8bfc1282f04942f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 18:19:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"430b194bff452326621da4602ab26ba570d54b06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:51:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76c96682eba4220a7dc9aed3b457151 : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5c3927cc03b44c09035fc54d3d673af : SUCCESS in 7m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c35c69741db345d39a6fcc8740d95df9 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36c2ad3f7d5a4aa8b459bf00e82a84fe : SUCCESS in 7m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c028e22f1b8a472aac6eb296fc63c54c : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59432ea3293d4b36aab165229b43ea8d : SUCCESS in 5m 04s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/4e44cdb1047e4f33af39ac47517dfd34 : SUCCESS in 57m 47s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/cf8ac1158f5e44e3a04bab8b8c5beff7 : SUCCESS in 1h 26m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ddc634e15a075afad1895da75aebde972b38c086","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:51:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2227c3ca403cf8bb5d3cc84a3467d8301fd0110f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:53:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71b6820d77a4ba18cdde24bf25a06b6 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"189588fec52690d0c61f70cd31ed2d8b8c72f4e1","revisions":{"189588fec52690d0c61f70cd31ed2d8b8c72f4e1":{"kind":"REWORK","_number":1,"created":"2019-10-15 00:16:55.000000000","uploader":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"ref":"refs/changes/85/688585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/85/688585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/85/688585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/85/688585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/85/688585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/85/688585/1"}}},"commit":{"parents":[{"commit":"2aaccf0da92169b705f149d9e91dc11d61dfd50a","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2aaccf0da92169b705f149d9e91dc11d61dfd50a"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-10-10 13:08:18.000000000","tz":120},"committer":{"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","date":"2019-10-15 00:16:55.000000000","tz":0},"subject":"bagpipe-bgp: fix a broken looking-glass path","message":"bagpipe-bgp: fix a broken looking-glass path\n\nlg.VALUE was used with something that isn\u0027t a directly\nusable value (that would be serializable), but a function\nthat we need to call first ; lg.SUBITEM is what needs to\nbe used in that case.\n\nChange-Id: Ic34aff147faac500c792146fa945f639c9c8992c\n(cherry picked from commit 80fceb3458720b3d7fdccfa9f4a59e1e370d0be7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/189588fec52690d0c61f70cd31ed2d8b8c72f4e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/189588fec52690d0c61f70cd31ed2d8b8c72f4e1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
