)]}'
{"id":"openstack%2Fnetworking-bagpipe~688654","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ie8d441b9d7669ec22236b2273d8e69a884751309","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/1848151","hashtags":[],"change_id":"Ie8d441b9d7669ec22236b2273d8e69a884751309","subject":"bagpipe-bgp: cleanly ignore RTC route of unsupported type","status":"MERGED","created":"2019-10-15 08:49:54.000000000","updated":"2019-11-04 20:28:39.000000000","submitted":"2019-11-04 20:27:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688654-1572899242815-aeac7fa8","meta_rev_id":"471ac48ed14aeed88d211db2bc1497ad166f5039","_number":688654,"virtual_id_number":688654,"owner":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-04 20:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2019-11-04 17:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2019-11-04 16:41:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":1,"date":"2019-11-04 19:06:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 16:41:20.000000000","updated_by":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"reviewer":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"state":"REVIEWER"},{"updated":"2019-11-04 20:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a11c4f2e72712494e22bd501d40a64d2dbd2ae8a","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-15 08:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e1122f91c4212a92d0437fb2f9a5bd218e51e44","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-15 08:50:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"48b1536538df35ca2a92ad4d4322cfce9785d1cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 10:39:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a9f7abe4fd141cdbf19d4d9f5c421e3 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00ff22030ede4d0eb7640e0ce5fc0cf8 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b411c4c50724e2abbad38be6ae35a3d : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f58f3589760403094826f1804ff2dd4 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d19f0f20f094d47b74c136930b36684 : SUCCESS in 10m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8983827d27be4cd6aef170c9dfc95015 : FAILURE in 7m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a19de1d8b1240bb9141a57506eccdee : SUCCESS in 8m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/66c9a5088e13456dbd336a77be2b77ed : SUCCESS in 1h 41m 32s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/57afbbc43e0a402b8588d7948cf3b46f : FAILURE in 21m 35s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c442b2c8be504c58b87285b4b7b78fae : SUCCESS in 1h 43m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2830465365056f65f782c920ce2b6c0d03525aad","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-15 12:09:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3cd7eeeaafcd0fd740ace4667ac8ef1b31ddc056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 14:01:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1fa671943dc4766b2565eec5d1859c4 : SUCCESS in 8m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2101ec44959c4e3abc82e286a2cb0a47 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/874ff378aa3044a99c06246ce2fe1684 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b88eece38ce245658a67907fa32b55cd : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55c1f5e29c341a9b264d67a98f7a828 : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48b398cd85204f5ba124f73c64885353 : FAILURE in 4m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5112366c70d4878a95d51b597294588 : FAILURE in 8m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/edc618225bd0470f9db02b676e2f73a7 : SUCCESS in 1h 35m 51s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/212a24071d554af29f090acac86f9421 : FAILURE in 19m 23s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c7b1d7044f7c49978db8874dc4ecbf14 : SUCCESS in 1h 45m 46s","accounts_in_message":[],"_revision_number":2},{"id":"d70fd37d9d39f325286379d72c5731cca7d561d6","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-18 08:41:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7efd93a72a8d097791aed027d4e9d391b266c5b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 10:11:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c3dd108c8c74b3abb33a947cd1d6c5a : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/545470ada2fb4d12afbeb439eb9e0b8c : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb5140e30891472bbf64ae3d8e53d51e : SUCCESS in 6m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5138e5217e2c4bae87bb158037a62468 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11eeecf9408e45db8318ddd56953029d : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7230d24b62d24668a417b427d0d8bff9 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d304c3dee0b4570bc65de7c6e932e34 : SUCCESS in 6m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/99f8aadc9dd3485a8ae6828af704156e : SUCCESS in 1h 26m 24s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f417c51dfb7343b9b3c9b7ff6a9fbba4 : FAILURE in 15m 46s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0387abb08e024dbaaea693082e52da2a : SUCCESS in 1h 10m 38s","accounts_in_message":[],"_revision_number":2},{"id":"2e3ace4edff1ff2d2ed75522bb31c15c28f49612","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-22 20:04:01.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 2f917d8efade8e0c7a0f9ec1a757d132dccaea8b","accounts_in_message":[],"_revision_number":2},{"id":"6aa41c4be49b93d930f3797b7bf2431f804580e1","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-22 20:04:29.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 58afc99155c0c00ebf9998d27b5b3430aef4d0f9","accounts_in_message":[],"_revision_number":2},{"id":"25e7ef4d8ae5017116671eba87de2dc04b1982ac","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-11-04 16:41:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"de4e2fcfe1dccbb6a48d9e790fc8903d1fb9a25a","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-11-04 17:15:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fa3ded2b3524ce715427674afff81779f6d7ee9d","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-11-04 19:06:01.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f36411eada10d432b830f218154560d76837e8cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 19:06:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4dfe5e7a077adc0a5f3a8e15caf2f81f42bdb00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 20:27:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419ee555f14b4443891b732dc55c9549 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85a15d9f425f4db7a7b43124b5c1e52b : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08b23b4016a1451b8ca9433e42f62481 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/225f7fc8ce484a24a1dce9046bc01a31 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e642d86467421991549248a6a15240 : SUCCESS in 7m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8e7f747b293f4b9689e1770ecc7dd64d : SUCCESS in 1h 20m 13s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b441d88315ff4634bc105118f209075d : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":2},{"id":"31aa3d51fa94f743b0cb788848b289c1b922c0cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 20:27:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cd8b2b0d11fbf5612ab5ef3671ae16993ff450bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 20:28:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72520bb7827d4a77a69418eba78930ea : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f4113085b069683f851e475c8bc738ce4d7478b2","revisions":{"7ba10bbd87c42664980d4b63cc2aecfa0058fd95":{"kind":"REWORK","_number":1,"created":"2019-10-15 08:49:54.000000000","uploader":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"ref":"refs/changes/54/688654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/54/688654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/688654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/688654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/688654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/688654/1"}}},"commit":{"parents":[{"commit":"a59b5c75751f5acf4e30223851ca1e6323c62e9a","subject":"Merge \"Force copy of dict items in \"for\" loop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a59b5c75751f5acf4e30223851ca1e6323c62e9a"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-10-15 08:46:54.000000000","tz":120},"committer":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-10-15 08:46:54.000000000","tz":120},"subject":"bagpipe-bgp: cleanly ignore RTC route of unsupported type","message":"bagpipe-bgp: cleanly ignore RTC route of unsupported type\n\nWhen a BGP peer advertises an RTC route (RFC4684) with\na route target of a type bagpipe-bgp does not support, we\nneed to ignore these cleanly, and not break and reset the\nBGP peering.\n\nNote that it is safe to plainly ignore such routes because,\nsince bagpipe-bgp does not support them, it will never need\nto propagate event for such RTs to a BGP peer.\n\nChange-Id: Ie8d441b9d7669ec22236b2273d8e69a884751309\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7ba10bbd87c42664980d4b63cc2aecfa0058fd95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7ba10bbd87c42664980d4b63cc2aecfa0058fd95"}]},"branch":"refs/heads/master"},"f4113085b069683f851e475c8bc738ce4d7478b2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-15 08:50:29.000000000","uploader":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"ref":"refs/changes/54/688654/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/54/688654/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/688654/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/688654/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/688654/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/688654/2"}}},"commit":{"parents":[{"commit":"a9f9f016afb67ac0a6038f4fd043e38cd9bd46a0","subject":"Merge \"bagpipe-bgp: fix a broken looking-glass path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a9f9f016afb67ac0a6038f4fd043e38cd9bd46a0"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-10-15 08:46:54.000000000","tz":120},"committer":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-10-15 08:50:19.000000000","tz":120},"subject":"bagpipe-bgp: cleanly ignore RTC route of unsupported type","message":"bagpipe-bgp: cleanly ignore RTC route of unsupported type\n\nWhen a BGP peer advertises an RTC route (RFC4684) with\na route target of a type bagpipe-bgp does not support, we\nneed to ignore these cleanly, and not break and reset the\nBGP peering.\n\nNote that it is safe to plainly ignore such routes because,\nsince bagpipe-bgp does not support them, it will never need\nto propagate event for such RTs to a BGP peer.\n\nChange-Id: Ie8d441b9d7669ec22236b2273d8e69a884751309\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f4113085b069683f851e475c8bc738ce4d7478b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f4113085b069683f851e475c8bc738ce4d7478b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
