)]}'
{"id":"openstack%2Fnetworking-bagpipe~688812","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Ftrain~Ia3df29060a6dfeb0f7b10769e8016762148a0a5b","project":"openstack/networking-bagpipe","branch":"stable/train","topic":"bug/1846767-stable/train","hashtags":[],"change_id":"Ia3df29060a6dfeb0f7b10769e8016762148a0a5b","subject":"Force copy of dict items in \"for\" loop","status":"ABANDONED","created":"2019-10-15 20:18:54.000000000","updated":"2019-10-17 06:29:21.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bac14601fc68de73af1bf28888621e84cce0da26","_number":688812,"virtual_id_number":688812,"owner":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-16 10:03:09.000000000","_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-15 22:55:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":-1,"date":"2019-10-16 08:12:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 22:55:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-16 08:12:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5d39c8840a40e6e2d878a3a1b0dfc491182fc83b","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-15 20:18:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1a78115660f0f9a00682a998a29472a2ba274bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:55:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f5cb205e3e4a4a832cdd5ed93712f3 : SUCCESS in 6m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f840334599147db99ab4427592bbd79 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e0fdaa076bb4905a428950bacb8f039 : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07838f29ac6844babb97afacd136c881 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40485cae598d4534a8adb17b81970d25 : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/523e57b33e144b26a9d1a7b57a1bed85 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdaca53328524c05b54df0525b781859 : SUCCESS in 7m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0defe968e30f4d93b0491daf99240965 : FAILURE in 31m 57s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/41ca8958aaff4b98ad8e928951921333 : FAILURE in 17m 58s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6033507a094a43cfb857d0b80e05d6a5 : SUCCESS in 2h 25m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3236c666f89d6b87b8f13fd44f7db2c20c6f4411","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-16 08:12:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52a5162fa0a0e1d0cb5eedfee4c43d1fec91117b","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-16 10:03:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c51792cf4e15128ebdf3473303ccda045f8ede3","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-17 06:29:21.000000000","message":"Abandoned\n\nWrapzed into 688811","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbdbb2a92c6b64d9c521fdc9247fcaa342c7f1f9","revisions":{"dbdbb2a92c6b64d9c521fdc9247fcaa342c7f1f9":{"kind":"REWORK","_number":1,"created":"2019-10-15 20:18:54.000000000","uploader":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"ref":"refs/changes/12/688812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/12/688812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/12/688812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/12/688812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/12/688812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/12/688812/1"}}},"commit":{"parents":[{"commit":"eb8f751a5cede7f8aa87139ff7d76ed9970facd6","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/eb8f751a5cede7f8aa87139ff7d76ed9970facd6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-07 09:38:37.000000000","tz":120},"committer":{"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","date":"2019-10-15 20:18:54.000000000","tz":0},"subject":"Force copy of dict items in \"for\" loop","message":"Force copy of dict items in \"for\" loop\n\nIn module bagpipe_bgp.common.dataplane_utils in\nObjectLifecycleManager.clear_objects() method there is for loop\nwhich iterates over items from dict.\nTo avoid changes of dict during iterations this patch forces always\nto create list which is copy of items from dict.\n\nChange-Id: Ia3df29060a6dfeb0f7b10769e8016762148a0a5b\nCloses-Bug: #1846767\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dbdbb2a92c6b64d9c521fdc9247fcaa342c7f1f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dbdbb2a92c6b64d9c521fdc9247fcaa342c7f1f9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
