)]}'
{"id":"openstack%2Fnetworking-bagpipe~689717","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fstein~I3c4662bf841f4129cedc3e23995f13ceac5431c0","project":"openstack/networking-bagpipe","branch":"stable/stein","hashtags":[],"change_id":"I3c4662bf841f4129cedc3e23995f13ceac5431c0","subject":"[stein-only] Fix neutron requirement constraint","status":"MERGED","created":"2019-10-21 11:13:30.000000000","updated":"2019-10-24 15:07:13.000000000","submitted":"2019-10-24 15:05:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689717-1571929552100-cd6aae0a","meta_rev_id":"3f58a9e09596074f7d66a7d1e3aa0e6ab4535607","_number":689717,"virtual_id_number":689717,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-21 12:09:34.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 15:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-10-22 15:50:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-10-24 12:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-10-24 13:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"date":"2019-10-22 15:50:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-10-24 13:27:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 15:50:20.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-10-22 15:50:30.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-10-24 12:01:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-10-24 13:27:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-24 15:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91a8d1cd061dd842bc78f5304764601d4e831a08","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-21 11:13:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ea54e6c382786b4f94fdcb5e3044755e3592bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 11:28:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b6eeb6bcfa48ea9861226caa8eaa89 : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1b3633529174e459a88eb089823952f : SUCCESS in 6m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef9ed24c2b3642e69bd819f70bc69974 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcdfb401be284748b501bb4fb974b374 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/893557c77c4045efa046759ef7df8f47 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f03faff0ab4d34bc9f2d19b72e9bb5 : SUCCESS in 4m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9dd41ece03084129bc523b1c5e7443d4 : FAILURE in 2m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7599f30bf0cf4cc6baf83eba698e24f3 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce44d4a8335c49d5aba75e4de14b1765 : SUCCESS in 6m 57s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/90f1857652214649a681738de8e9a551 : FAILURE in 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4143a0102f022f848deb2fc34a3bb80f6cff9947","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-21 11:34:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b44d7110762f2c7a2dfa6b2287f3bae0af90c7e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-21 12:09:34.000000000","message":"Patch Set 2:\n\nexample for stable periodic tox-py27 failing job:\n\nhttps://zuul.opendev.org/t/openstack/build/141941877bb544e0bfa520af5996d1ba","accounts_in_message":[],"_revision_number":2},{"id":"6c382a9f01ea315b52ff20051d3284d55141d205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 13:03:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebf37478b7f54308a046e0bb83a36703 : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a960e5d4735249a5bc23169598a41ea9 : SUCCESS in 6m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/450865a286eb44de94b0384f89edb392 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a61b3801b52d4c2386ba90a1e5503e2c : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e39a93f473143ba837e0247036e625b : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7d5c6af8f524e3b9dd986089a912bd8 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/274991479e47431f9c87319fecfbc6cd : SUCCESS in 4m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/16b71f082b734d3c8493d65e9e2d0406 : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/059e895ff4844eb1a79ca3319f52ffad : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/900195684f794973a4143b1931f80456 : SUCCESS in 6m 36s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/09beb774a853440c99a4675b13f1bf24 : SUCCESS in 53m 53s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/68ef6ec850b548a685133d393cf75752 : SUCCESS in 1h 23m 25s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/5ae0a1eaabe2406185f2bf5a87f5f7c6 : FAILURE in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"03af9bb85cc0940975b6cbee7a44146027b748cc","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-22 15:50:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2d6130e0692a47e1769b5f0b2bd3dae01ff2afad","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-24 12:01:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a8c56014fac9a38bf97f369784499e0bacaf66fe","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-24 13:27:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81e6166148088a25891639ad5e38b90ccfbe4da5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:27:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"827554a468d3bdee3a15f3ff0a1eff89ec391519","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 15:05:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03a65465b287495097113bceeca45066 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b9c2f7eea1f4977bd420b797e45ef11 : SUCCESS in 6m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/40af104580da47d7870e1c08ae35fb4c : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a99e80f91d2494f9e296c33520dca3b : SUCCESS in 8m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2b5f0d9828c40e8b876456b2e8a3ad4 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/540647c1571241db8481ef98f8780635 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e3ac108056d48fabc86456bfcd16790 : SUCCESS in 6m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e0de90fa9a44758833759414fa9425f : SUCCESS in 3m 48s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/7dc41bedcf2b466cabb36571295bb0db : SUCCESS in 1h 09m 28s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/76b26000498447e1a9cf06e81c479611 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":2},{"id":"fe878b294f87e86b493ae1aaf245cca4c104f6fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 15:05:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3f58a9e09596074f7d66a7d1e3aa0e6ab4535607","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 15:07:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef583bfb124248ec977d9c8f434dc3e9 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/327a444c5cb1434db20e4b227921064d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4549dd2a1d93620692fc2756237e18c50a7d1b7e","revisions":{"539c6139768180a809a3af866831a4ced2cf7864":{"kind":"REWORK","_number":1,"created":"2019-10-21 11:13:30.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/17/689717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/689717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/689717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/689717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/689717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/689717/1"}}},"commit":{"parents":[{"commit":"2aaccf0da92169b705f149d9e91dc11d61dfd50a","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2aaccf0da92169b705f149d9e91dc11d61dfd50a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2019-10-21 11:04:10.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2019-10-21 11:12:11.000000000","tz":120},"subject":"[stein-only] Fix neutron requirement constraint","message":"[stein-only] Fix neutron requirement constraint\n\nTrain version of neutron (15.0.0) was released and since then\ntest jobs fail with:\n\nContextualVersionConflict: (openstacksdk 0.27.0 (/home/zuul/src/opendev.org/openstack/networking-bagpipe/.tox/py27/lib/python2.7/site-packages), Requirement.parse(\u0027openstacksdk\u003e\u003d0.31.2\u0027), set([\u0027neutron\u0027]))\n\nHowever jobs pick up the latest neutron from stable/stein, the\n\"Run tox without tests\" or the local tests fail.\n\nChange-Id: I3c4662bf841f4129cedc3e23995f13ceac5431c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/539c6139768180a809a3af866831a4ced2cf7864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/539c6139768180a809a3af866831a4ced2cf7864"}]},"branch":"refs/heads/stable/stein"},"4549dd2a1d93620692fc2756237e18c50a7d1b7e":{"kind":"REWORK","_number":2,"created":"2019-10-21 11:34:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/17/689717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/689717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/689717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/689717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/689717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/689717/2"}}},"commit":{"parents":[{"commit":"2aaccf0da92169b705f149d9e91dc11d61dfd50a","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2aaccf0da92169b705f149d9e91dc11d61dfd50a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2019-10-21 11:04:10.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2019-10-21 11:33:20.000000000","tz":120},"subject":"[stein-only] Fix neutron requirement constraint","message":"[stein-only] Fix neutron requirement constraint\n\nTrain version of neutron (15.0.0) was released and since then\ntest jobs fail with:\n\nContextualVersionConflict: (openstacksdk 0.27.0 (/home/zuul/src/opendev.org/openstack/networking-bagpipe/.tox/py27/lib/python2.7/site-packages), Requirement.parse(\u0027openstacksdk\u003e\u003d0.31.2\u0027), set([\u0027neutron\u0027]))\n\nHowever jobs pick up the latest neutron from stable/stein, the\n\"Run tox without tests\" or the local tests fail.\n\nDue to touching the requirements.txt the requirement check job also\ncalled, which shows that Sphinx requirement has to be aligned, too.\n\nChange-Id: I3c4662bf841f4129cedc3e23995f13ceac5431c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4549dd2a1d93620692fc2756237e18c50a7d1b7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4549dd2a1d93620692fc2756237e18c50a7d1b7e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
