)]}'
{"id":"openstack%2Fnetworking-bagpipe~690395","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Ftrain~Ie8d441b9d7669ec22236b2273d8e69a884751309","project":"openstack/networking-bagpipe","branch":"stable/train","topic":"bug/1848151-stable/train","hashtags":[],"change_id":"Ie8d441b9d7669ec22236b2273d8e69a884751309","subject":"bagpipe-bgp: cleanly ignore RTC route of unsupported type","status":"MERGED","created":"2019-10-22 20:04:01.000000000","updated":"2019-11-22 15:44:19.000000000","submitted":"2019-11-22 15:43:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690395-1574437389098-752aeac9","meta_rev_id":"e1926cbb4b4e71e39810e469052abe853959267e","_number":690395,"virtual_id_number":690395,"owner":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-11-22 12:23:40.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 15:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-11-18 09:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2019-11-18 09:37:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 14:24:05.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-11-05 03:38:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-18 09:37:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-11-22 12:23:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-11-22 15:43:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc95338c1b56d9fbd6044e253fba9feacf4624ed","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-10-22 20:04:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"92e11dbc82fa38fa63782ad70eed46821ffae372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 21:40:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ed9efbf5704bafb27a5db778367241 : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7315f807c8240079de50b0ecd9e4711 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec10faafdb5741bc82b8604124a96518 : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4cfdd1296e9443a383d6cc1c2c4226b2 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b7ce7482c0648748b6c62f81368d55f : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe4500743ae84594931c5292fc18172d : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dc4e836797f488486b003cc3896864b : SUCCESS in 7m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/99b9ec4a83884d41976724c3ce6fd58d : SUCCESS in 1h 30m 16s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/3affc8db81fb455581c5466c6ae2cc49 : FAILURE in 17m 25s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ad552e50596748cf8512426f4119cf45 : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4f9df31c6c69a3a67cb14c658234e31d350559fb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-25 14:42:23.000000000","message":"Patch Set 1:\n\nWe need to wait for the change on master to merge before approving this and stable/stein version.","accounts_in_message":[],"_revision_number":1},{"id":"898a977a64ac9fb747f01e84532db4feaab1ffb2","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-11-04 20:44:18.000000000","message":"Patch Set 1:\n\nThe fix has now merged in master.","accounts_in_message":[],"_revision_number":1},{"id":"d34fcb18947e65aa5fd1f6bc64ddde14ac1b9d47","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-05 03:38:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a66930356d474df520837c3a03db328834b0fe33","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-18 09:36:50.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1999e55b34f3d1d3f254d4878d0a2359993863d1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-18 09:37:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI just added info about from where it was cherry-picked.","accounts_in_message":[],"_revision_number":2},{"id":"f43d8063dc9862202713f1d9adc4dd5b38cc7923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 11:15:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451366f8f1ba47fd9f0b19bb886d4692 : SUCCESS in 7m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a37a8aab04244ef825dd112f3264d84 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3d810d855ff41229b53ea05612f2566 : SUCCESS in 7m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/304232de65a54609bba75b690841cd88 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7bbbe3b45854cc38397ced144958d81 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f48aa9b1762d48d3ba416764657cfa52 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf742e42046646db9a24d8367b99d0c6 : FAILURE in 6m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/6cc660666d8941c3903acba8336247ad : SUCCESS in 1h 34m 09s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/30a9d842d5104c3fb8a2edc6868bd8f5 : FAILURE in 17m 21s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/669e88057f654430a6e2857a1972ae38 : SUCCESS in 1h 25m 07s","accounts_in_message":[],"_revision_number":2},{"id":"046a026bc372fcd5d12c9644a8d1bfdc5f973888","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-22 12:23:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"97d269ba7b6c7eada864d006c9a9380bd3f278b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 14:02:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e84a26bfa304a87ae54a76e9a1327f4 : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba795273ad1141f397899df6bb53b14a : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b6d54426cd6444cacc5f93b0f77cbd7 : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58a0b21075324c5f99ad1e726305ec63 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24117e90027b429fbf58cb0dc8d9a63c : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e189e7fc9a9143cf8eff74c83dfafd8b : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22eace99283847c28ff4bb118da1c159 : SUCCESS in 8m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/d15bb7be5c544c6f804c19933098ef41 : SUCCESS in 1h 34m 59s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/8aab373146c8436db2b35d98c1d49e06 : FAILURE in 18m 10s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/e7bf654978d346fb932dd19dda73e9ba : SUCCESS in 1h 34m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d5a1d67905f2e00675f9fa80e158f2849cdade15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 14:02:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c61294bd32afb7fda30d5bcb4c39bfcb3f0f35fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 15:43:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c498dc736c4bbfa4434f0d44234c41 : SUCCESS in 9m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/60f16b7114284b8faad22ae905b24634 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/443af967b8a149bf848edca8dd89834d : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d61e29a01e84b20a8752040970e3578 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d306b124ea1e42bbb8681e9b4ebe9f61 : SUCCESS in 5m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/331fb0ec58b84741a0c1a6d8df2d067e : SUCCESS in 1h 35m 04s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3dc9fc7edc7244eb83ec12fa7e853363 : SUCCESS in 1h 15m 33s","accounts_in_message":[],"_revision_number":2},{"id":"bb011bd6e15e4a9fb06965ec18b5bcd41ef1c7ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 15:43:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e5d1a812cf9304d8849abc3da0acf62e8dce618f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 15:44:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c4cee20c974c6b8d963c8d67d12f4a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a61ad6c5d4eb6c6c90fa4f0aa2cd44acaa6da56e","revisions":{"2f917d8efade8e0c7a0f9ec1a757d132dccaea8b":{"kind":"REWORK","_number":1,"created":"2019-10-22 20:04:01.000000000","uploader":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"ref":"refs/changes/95/690395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/690395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/690395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/690395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/690395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/690395/1"}}},"commit":{"parents":[{"commit":"626d4a07c117e054eff6c9d73a6acbd8791dee18","subject":"assorted fixes in networking_bagpipe.bagpipe_bgp module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/626d4a07c117e054eff6c9d73a6acbd8791dee18"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-10-15 08:46:54.000000000","tz":120},"committer":{"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","date":"2019-10-22 20:04:01.000000000","tz":0},"subject":"bagpipe-bgp: cleanly ignore RTC route of unsupported type","message":"bagpipe-bgp: cleanly ignore RTC route of unsupported type\n\nWhen a BGP peer advertises an RTC route (RFC4684) with\na route target of a type bagpipe-bgp does not support, we\nneed to ignore these cleanly, and not break and reset the\nBGP peering.\n\nNote that it is safe to plainly ignore such routes because,\nsince bagpipe-bgp does not support them, it will never need\nto propagate event for such RTs to a BGP peer.\n\nChange-Id: Ie8d441b9d7669ec22236b2273d8e69a884751309\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2f917d8efade8e0c7a0f9ec1a757d132dccaea8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2f917d8efade8e0c7a0f9ec1a757d132dccaea8b"}]},"branch":"refs/heads/stable/train"},"a61ad6c5d4eb6c6c90fa4f0aa2cd44acaa6da56e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-18 09:36:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/690395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/690395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/690395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/690395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/690395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/690395/2"}}},"commit":{"parents":[{"commit":"626d4a07c117e054eff6c9d73a6acbd8791dee18","subject":"assorted fixes in networking_bagpipe.bagpipe_bgp module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/626d4a07c117e054eff6c9d73a6acbd8791dee18"}]}],"author":{"name":"Thomas Morin","email":"thomas.morin@orange.com","date":"2019-10-15 08:46:54.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-18 09:36:28.000000000","tz":0},"subject":"bagpipe-bgp: cleanly ignore RTC route of unsupported type","message":"bagpipe-bgp: cleanly ignore RTC route of unsupported type\n\nWhen a BGP peer advertises an RTC route (RFC4684) with\na route target of a type bagpipe-bgp does not support, we\nneed to ignore these cleanly, and not break and reset the\nBGP peering.\n\nNote that it is safe to plainly ignore such routes because,\nsince bagpipe-bgp does not support them, it will never need\nto propagate event for such RTs to a BGP peer.\n\nChange-Id: Ie8d441b9d7669ec22236b2273d8e69a884751309\n(cherry picked from commit f4113085b069683f851e475c8bc738ce4d7478b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a61ad6c5d4eb6c6c90fa4f0aa2cd44acaa6da56e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a61ad6c5d4eb6c6c90fa4f0aa2cd44acaa6da56e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
