)]}'
{"id":"openstack%2Fnetworking-bagpipe~700534","triplet_id":"openstack%2Fnetworking-bagpipe~master~I24dde51e2bb652f20c6b19c65e039616457928bb","project":"openstack/networking-bagpipe","branch":"master","topic":"fixwords","hashtags":[],"change_id":"I24dde51e2bb652f20c6b19c65e039616457928bb","subject":"Fix the url errors","status":"MERGED","created":"2019-12-25 03:19:51.000000000","updated":"2020-12-10 10:07:47.000000000","submitted":"2020-12-10 10:06:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700534-fixwords","meta_rev_id":"9623e83094ee05da34338c5a0fcbbff0abf18024","_number":700534,"virtual_id_number":700534,"owner":{"_account_id":27399,"name":"zhouguowei","email":"zhouguowei@inspur.com","username":"zhouguowei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-10 10:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2020-12-10 08:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2020-12-10 08:37:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-25 05:27:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 08:37:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"97875330a8c50c4f0d6229b2816881e54c85df1d","author":{"_account_id":27399,"name":"zhouguowei","email":"zhouguowei@inspur.com","username":"zhouguowei"},"date":"2019-12-25 03:19:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5120ab7ca80efcac9f81ebd5f680a253073d4c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-25 05:27:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e74d561fd6884f9fb91752c506a4248a : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/516e7945678d4480a5926a52bfc248cd : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/708765895f8a461ba04505b83d0da619 : SUCCESS in 6m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7d56ce608515480ca98b556fdb18746b : SUCCESS in 1h 36m 29s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/cab829ec07ab4e44b1c2d1181b1900ea : FAILURE in 14m 58s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/bff4b1b7c5c84b11802a67e07eb1b58e : SUCCESS in 1h 17m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2c303c5409b140c26e5e80fc08efa87cead77b45","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-10 08:37:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial change so I will quickly approve it","accounts_in_message":[],"_revision_number":1},{"id":"303897686bc3248d3875570ea497021d93940ee7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 08:38:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c7df1700c0cef5f8651c224d2612931d6169cb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 10:06:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b3e39d2727b04a1dbbdc49a3054efef7 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419b3ac754dc4b32b75fe7d9cc0a1066 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6867162d98664927b1166c7e3151604c : SUCCESS in 6m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/23abd662ceff4ceb8eef63b040fb2bdf : SUCCESS in 1h 19m 05s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/d2ca582ddac14f39a937545d6b30083a : SUCCESS in 1h 14m 21s","accounts_in_message":[],"_revision_number":1},{"id":"607e9caf5e4262ef39d118b392ba377897075f1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 10:06:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9623e83094ee05da34338c5a0fcbbff0abf18024","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 10:07:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3093c2e298443b88eca86d4b9fc2064 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0c75a922f54261d85457b070a88e920d4269abb","revisions":{"f0c75a922f54261d85457b070a88e920d4269abb":{"kind":"REWORK","_number":1,"created":"2019-12-25 03:19:51.000000000","uploader":{"_account_id":27399,"name":"zhouguowei","email":"zhouguowei@inspur.com","username":"zhouguowei"},"ref":"refs/changes/34/700534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/34/700534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/700534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/700534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/700534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/700534/1"}}},"commit":{"parents":[{"commit":"410cbb01c1a6d7d392b1a1b322b14126aaf2be29","subject":"use object common_types from neutron-lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/410cbb01c1a6d7d392b1a1b322b14126aaf2be29"}]}],"author":{"name":"zhouguowei","email":"zhouguowei@inspur.com","date":"2019-12-25 03:17:56.000000000","tz":480},"committer":{"name":"zhouguowei","email":"zhouguowei@inspur.com","date":"2019-12-25 03:19:31.000000000","tz":480},"subject":"Fix the url errors","message":"Fix the url errors\n\nChange-Id: I24dde51e2bb652f20c6b19c65e039616457928bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f0c75a922f54261d85457b070a88e920d4269abb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f0c75a922f54261d85457b070a88e920d4269abb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
