)]}'
{"id":"openstack%2Fnetworking-bagpipe~702208","triplet_id":"openstack%2Fnetworking-bagpipe~master~Iac22d787af46ad57f24391eae2188b0c74c84299","project":"openstack/networking-bagpipe","branch":"master","topic":"unittest2","hashtags":[],"change_id":"Iac22d787af46ad57f24391eae2188b0c74c84299","subject":"Remove references for unittest2","status":"MERGED","created":"2020-01-13 12:23:54.000000000","updated":"2020-01-16 19:04:48.000000000","submitted":"2020-01-16 19:03:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702208-1579201382893-d94f54cc","meta_rev_id":"7c0b662dd922515996f12918d360f4130a2cbcfd","_number":702208,"virtual_id_number":702208,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 19:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2020-01-16 16:57:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":1,"date":"2020-01-16 16:57:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 16:57:37.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-01-16 19:03:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87ba5c7cc9ce9bafb061657870701fb33965ac2b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-13 12:23:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdd9299944928b51392bc1f7519d5f3971f15620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 12:59:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc371a8be83d48e4942de6577a52f22f : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d80c0d7b09a4b6d855a9f7b6569ece4 : FAILURE in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2371775a56fb47e797eccf8726490718 : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a11d10bc7e304a588f739376c75f1785 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8eaf41f231c4149bf5934d6d7d25837 : FAILURE in 4m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/686270833cc0438f8bce6167b7e26ebe : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d79b4f958be4005bae35210a467e8e9 : FAILURE in 4m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64cf27e2c27f44eeb41eba46d150006f : SUCCESS in 9m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d76c9165a1814403a79d119a1e80bb87 : FAILURE in 13m 32s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/edea7a58995e45a499749a182ec602c7 : FAILURE in 5m 26s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0ba9896a0b844c18b1f1563e7041b918 : FAILURE in 9m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ae9204d19f82e98884bb5ba1c5dae4e1e4308cf9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-14 10:00:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"320303dff88d1b71f3425769dc7e89f8008bdf05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:00:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe23421d8dd7445397827aeef13cb407 : SUCCESS in 7m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8da22e2476e451890f346b0851ad5be : FAILURE in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea6a25f17f12482e83c40954b168c795 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2bd2147ed124b649dc0c911bdb5b5fc : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa787f915bc4c5f830f8357e59a67a2 : SUCCESS in 6m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a9bf00ce2ec4ec09cb98928529adb8b : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b0dc35e42184f74a3080ed130e6e8cf : FAILURE in 7m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c985605cef0f49f99f4ebd51eb2d56b7 : SUCCESS in 9m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6dd1f26020814b8fbd581bf65ed8b397 : SUCCESS in 1h 52m 55s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2f08668c26a843c881e746f2aa6910c4 : FAILURE in 21m 55s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/07a623f7815d47a39a18030d4c229565 : SUCCESS in 1h 33m 20s","accounts_in_message":[],"_revision_number":1},{"id":"737badf75c65e4c8b15c0289cf757199175c8328","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-15 11:57:39.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6f333fad008a1a675fffe9a16dd7281f006ba20f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 13:48:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b97b99f386d84d97a6248e391cf6e5f1 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85b5fff310c0419fbc6952f252bcaa24 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55692a59aabf454cb41c289748f14027 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728dd5250a5b47ffa06d436b8eda945a : SUCCESS in 7m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf1216b1df76428c98eb1a8bd33f6c27 : SUCCESS in 3m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/418a63803fbf47e2963d3b8f534f7f73 : FAILURE in 4m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f8053f326ed478ab6660248ef2b59ff : SUCCESS in 8m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a64123c555df425ba638e44a76c514c5 : SUCCESS in 1h 46m 06s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/71fba2bcf88d450da8b47fbc682a669d : FAILURE in 24m 41s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f1519278ec6848fb88c2e2a2e16176a0 : SUCCESS in 1h 30m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2121a11b7853f42c5ed9ad6ccd6ac3e09b525e22","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-01-16 16:57:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1a28007706a6cd11862ae35f3035a1d8eff85ece","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 16:57:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4b9faeb4743b16e5aeebce0b8f88d5bedc3c12c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 19:03:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5f558bd14d4778b756f6e40933a9fe1c25760f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 19:03:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5057daae96f74518b5c272bf3b93b620 : SUCCESS in 9m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87381a27a75f4d5abf84c302ba8db09a : SUCCESS in 10m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18e0ab070f9741a7af594f658b99f4c0 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61a51799b56b4beca3c16978464642c9 : SUCCESS in 6m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/96b717cde45741cfa4adbb11be0440dc : SUCCESS in 4m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/254aa180c1c84921bba66c116044aadd : SUCCESS in 2h 01m 02s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/849f492457414ee79608c589da72d784 : SUCCESS in 1h 46m 29s","accounts_in_message":[],"_revision_number":2},{"id":"27f85bb058cf85a743e71787dda033886a1632d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 19:04:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fde046bbfc34984bc9188f48a50d5ad : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f0c6cc49d585d21de8be882f4f2d5649cd833a07","revisions":{"6cfadd001b933b683812f564a214b5702416746e":{"kind":"REWORK","_number":1,"created":"2020-01-13 12:23:54.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/08/702208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/08/702208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/702208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/702208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/702208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/702208/1"}}},"commit":{"parents":[{"commit":"873e6725274cbbfaef0c2e7eeb0de4e3d92c18e4","subject":"Update and replace http with https for doc links","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/873e6725274cbbfaef0c2e7eeb0de4e3d92c18e4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-13 12:22:59.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-13 12:22:59.000000000","tz":60},"subject":"Remove references for unittest2","message":"Remove references for unittest2\n\nunittest2 was needed for python\u003c\u003d2.6, it is time to remove it from\nrequirements.\n\nChange-Id: Iac22d787af46ad57f24391eae2188b0c74c84299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6cfadd001b933b683812f564a214b5702416746e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6cfadd001b933b683812f564a214b5702416746e"}]},"branch":"refs/heads/master"},"f0c6cc49d585d21de8be882f4f2d5649cd833a07":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-15 11:57:39.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/08/702208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/08/702208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/702208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/702208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/702208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/08/702208/2"}}},"commit":{"parents":[{"commit":"46b433cac609b3fc9796217c6cb0ee1d86b80fca","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46b433cac609b3fc9796217c6cb0ee1d86b80fca"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-13 12:22:59.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 11:56:58.000000000","tz":60},"subject":"Remove references for unittest2","message":"Remove references for unittest2\n\nunittest2 was needed for python\u003c\u003d2.6, it is time to remove it from\nrequirements.\n\nChange-Id: Iac22d787af46ad57f24391eae2188b0c74c84299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f0c6cc49d585d21de8be882f4f2d5649cd833a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f0c6cc49d585d21de8be882f4f2d5649cd833a07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
