)]}'
{"id":"openstack%2Fnetworking-bagpipe~702895","triplet_id":"openstack%2Fnetworking-bagpipe~master~I627217c439dd0c90361580956c9ca13480fc10fe","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/1859861","hashtags":[],"change_id":"I627217c439dd0c90361580956c9ca13480fc10fe","subject":"Fix fullstack tests","status":"ABANDONED","created":"2020-01-16 15:33:40.000000000","updated":"2021-04-14 13:53:15.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"91e03ad82d20da449dd56a2cba0ba5e953a4f5ec","_number":702895,"virtual_id_number":702895,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-13 12:52:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 16:12:23.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-03-31 20:07:42.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-04-01 21:44:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-04-06 09:52:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-10-13 12:52:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b9fbd6d3fb21b67e8e0c0ddb8733fdfc10ae599","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-16 15:33:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d3fe04d9a73a2dd9a344e7bc90296fa8382d201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 17:01:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/005f2d820dbb48e6afe455c530b77bc1 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b368c36d4c814a80b453881e9eab9ac3 : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a4bf0dc4944448a9a40accce3f1567f : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/621b348713514674bb9f911739ded808 : SUCCESS in 12m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efd347cb3fc041fdae9e8d7e135a1742 : FAILURE in 10m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce581bb0af88487aa8631c2ebf527f7a : SUCCESS in 12m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/48324bc1b92e45c4a2a2c08ba7833779 : SUCCESS in 1h 24m 56s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0c12925ccb2f4221a8195ecc6159f6f7 : FAILURE in 20m 03s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b051f0b8f0104f668e878cb47d3df0c8 : SUCCESS in 1h 21m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ef525f7f76016b8e556616f3ccc37772f3c25fd8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-17 09:13:03.000000000","message":"Patch Set 1: Workflow-1\n\nSome config option seems to be still wrong...","accounts_in_message":[],"_revision_number":1},{"id":"567011efffd1a704eff75c287bb4507e61ccbf48","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-17 15:38:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"39df08940f7c24c7f92aa1d828900daec56f7c9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 18:44:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa60b6ce8fc4160b4e395dff95632af : SUCCESS in 11m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7a325b9d7ae4ce5b3501d7b69d27a46 : SUCCESS in 9m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b74ba93de2b947c08247014737672194 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce37075076c438d948d56bd9c9b62a4 : SUCCESS in 10m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/741baf6c67634785a0364bc9da5e318f : FAILURE in 6m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57563d267ed74c4a8e5b12b479682bfb : SUCCESS in 10m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3bd7f37c26094a0ebc13201644e5440e : RETRY_LIMIT in 20m 22s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ad3c06a744ef4578804f93cb6534e81e : FAILURE in 31m 32s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/40c554554591409b9742a959bbd69733 : RETRY_LIMIT in 21m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3aa896299a74d1a3ff3e3239cd93470f3e676277","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-20 15:46:54.000000000","message":"Patch Set 2:\n\nHmmmm....\nDuring test runs in the bagpipe-bgp logs I see tracebacks like this:\n\n2020-01-17 16:07:04.285 25525 INFO networking_bagpipe.bagpipe_bgp.vpn.ipvpn.2 [-] method VRF.vif_unplugged called with (\u0027fa:16:3e:a4:6c:49\u0027, \u002720.0.0.189\u0027) {}\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers [-] detach_localport: An error occurred during local port unplug from VPN: KeyError: (\u0027fa:16:3e:a4:6c:49\u0027, \u002720.0.0.189\u0027)\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers Traceback (most recent call last):\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers   File \"/opt/stack/new/networking-bagpipe/networking_bagpipe/bagpipe_bgp/api/controllers.py\", line 248, in process\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers     self.manager.unplug_vif_from_vpn(**detach_params)\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers   File \"/opt/stack/new/networking-bagpipe/networking_bagpipe/bagpipe_bgp/common/log_decorator.py\", line 39, in wrapper\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers     return method(*args, **kwargs)\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers   File \"/opt/stack/new/networking-bagpipe/networking_bagpipe/bagpipe_bgp/vpn/manager.py\", line 410, in unplug_vif_from_vpn\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers     vpn_instance.vif_unplugged(mac_address, ip_address_prefix)\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers   File \"/opt/stack/new/networking-bagpipe/networking_bagpipe/bagpipe_bgp/common/log_decorator.py\", line 39, in wrapper\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers     return method(*args, **kwargs)\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers   File \"/opt/stack/new/networking-bagpipe/networking_bagpipe/bagpipe_bgp/vpn/ipvpn/__init__.py\", line 363, in vif_unplugged\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers     direction \u003d self.endpoint_2_direction[endpoint]\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers KeyError: (\u0027fa:16:3e:a4:6c:49\u0027, \u002720.0.0.189\u0027)\n2020-01-17 16:07:04.285 25525 ERROR networking_bagpipe.bagpipe_bgp.api.controllers \n\nsee: https://8e8576cef9278262b7b7-2de800c51e9ddb963f260573366c025b.ssl.cf5.rackcdn.com/702895/2/check/legacy-networking-bagpipe-dsvm-fullstack/ad3c06a/logs/dsvm-fullstack-logs/TestConnectivitySameBGPVPN.test_l3_network_connectivity_Linuxbridge_/","accounts_in_message":[],"_revision_number":2},{"id":"fbb98069554cf2f51e768017bd3ae8f4b2d0c413","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-01-22 16:12:23.000000000","message":"Patch Set 2:\n\n\u003e Hmmmm....\n \u003e During test runs in the bagpipe-bgp logs I see tracebacks like\n \u003e this:\n \u003e \n...\n \u003e \n \u003e see: https://8e8576cef9278262b7b7-2de800c51e9ddb963f260573366c025b.ssl.cf5.rackcdn.com/702895/2/check/legacy-networking-bagpipe-dsvm-fullstack/ad3c06a/logs/dsvm-fullstack-logs/TestConnectivitySameBGPVPN.test_l3_network_connectivity_Linuxbridge_/\n\n\nThis seems like a bug in the cleanup test ; perhaps does it happen when a delete is done on an endpoint for which attach failed or is still in progress.\n\nI would not expect this to block the test though (but perhaps I\u0027m wrong).\n\n\nOne thing about fullstack tests: they are \"nice to have\" because they test a log of things with few resources, but they were quite a huge task to write, and to maintain. See https://review.opendev.org/#/c/500109/ which I abandoned for lack of time to find a fix: perhaps the technical context has changed and this is doable now.   My point is: don\u0027t spend too much energy on this one.   An alternative to consider would be to write multinode devstack jobs (if possible covering the OVS and linuxbridge drivers, but this is not a prerequisite).","accounts_in_message":[],"_revision_number":2},{"id":"924799af4dfd44dbfa68c15ed174dfab663ac55d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-14 16:11:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7b268aea868994a675c306ad8a9ee08477496165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:42:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374ad27a66ab4d1399b2dde347f54140 : SUCCESS in 11m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b785e4028ae450ba7ddc4d0c126afcc : SUCCESS in 11m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4561dcd1e9d848e3855a4564753eafc4 : SUCCESS in 11m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e07b08110684bd39c2395a37bfeb97f : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06807f1837e24f67a08c19363b0ea588 : SUCCESS in 11m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ded5d0a00cef4a36a6fd1d4c850e0ac1 : FAILURE in 8m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3df35453237e401c8c401228fdbb57d6 : SUCCESS in 10m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/96e42354d1194959a6215d7a4d2d80ef : SUCCESS in 1h 28m 10s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0454548879134f3a984928c165d3cb05 : FAILURE in 26m 44s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/d1aaa59afeb545bd9d14d314d83453ee : SUCCESS in 1h 17m 12s","accounts_in_message":[],"_revision_number":2},{"id":"678292eedae70fe1b2575a19e27dc9810eb75d68","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-16 20:59:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a68617f76ad5236835116340da13ce49eedd0fa4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-16 21:00:56.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"2a0f71502723701ef3b4dcd18eeb24b81b3d8682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 21:00:56.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"260d7a8252c72629b5302eb8ad333b0a5e1bbaa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 22:48:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7e1ea44270749e398973b286b8056dd : RETRY_LIMIT in 2m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/807e03b98d954b70923c183e1c17cf7f : RETRY_LIMIT in 2m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ee8ef2f37ef4beba9dc27f2aca52f8c : RETRY_LIMIT in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b37b592f1034840804408a646f37802 : RETRY_LIMIT in 2m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ba34e541da492d90087a504084d79e : RETRY_LIMIT in 2m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a291969de374301b05ac63c1de0b112 : RETRY_LIMIT in 2m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc88caf1523b411ca66762779733e59d : RETRY_LIMIT in 2m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/364032eeeb9d448cb0525cec52f7c043 : RETRY_LIMIT in 2m 24s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c73705bc663c42829d0e3df6e5ef915d : SUCCESS in 1h 29m 05s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0d9c5d2776db45f4afc69c6cdd28f4f2 : FAILURE in 22m 06s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3a891b5cfcee4c75bfdded97ef45e43f : SUCCESS in 1h 37m 06s","accounts_in_message":[],"_revision_number":4},{"id":"928e480fb457db4054bd1d967fff4f24b6d08a1d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-17 08:49:58.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"381b55a1ba2b6f5e4f419d5cdff226f7af0b1697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 10:27:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc88cffa3d0c40e18d2e2209abf199c0 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d051e4f68ffa4203b964b9d0e86dbeab : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e609a9a45aa740628cb79dc06dac9114 : FAILURE in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cd7bff8f7a34687a52ba02ba154314d : FAILURE in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e5df87165540c0aea362ae8b2a2f48 : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9900e8e31ff347f9b411f43245242269 : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dbaaaae40e740fab3dc14decdfe2d8e : FAILURE in 8m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9645bc6179fa4456a45d631609a92598 : SUCCESS in 8m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/bd78252ac393471d980f8e7927e88363 : SUCCESS in 1h 29m 51s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/af474767ff4942debb6e86f268dcfece : FAILURE in 19m 48s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f8dadcc715294705b60a728cdc60f12c : SUCCESS in 1h 32m 16s","accounts_in_message":[],"_revision_number":5},{"id":"a6715b29cec8959179e653d92ff35ed87f35e2d6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-01 21:44:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"097cb1eeeb936b13e6d684fffda3c4612da432b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 23:22:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda979c916da425d9a89726a85f0f97b : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c10cce9b070840c58b22184f020afff2 : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2e5e06da0684dc59fdf395dde3002e5 : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c63ee5cfbfc24f85a98277af132e6735 : FAILURE in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/997e2d361e4b4f56a4f915d3feac8090 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/495301294b954a8fb935a276ff71c417 : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2de7ae8bd4834e09a12c5fde6b1957a0 : FAILURE in 5m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78506c76034b4c5ea167be77b48dc261 : SUCCESS in 7m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/98375128ef5b4c1383822acff6de7520 : SUCCESS in 1h 28m 50s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/eab5fca26dff4d0ab5df94deba5c0655 : FAILURE in 26m 32s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6d0be19a042143bead5e2b3cab295c69 : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":6},{"id":"81afbec6a6c4b58a06655dcd537880f48c9698f2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-02 12:16:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ad3db9473c0a473bc8fa23a3bd691b1b4133c150","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-03 07:58:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fbbfe1e88aca4c82be354bb397cd3f17712b1931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 09:33:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a58e5ded784fe59e5ee7ae9b5688e0 : SUCCESS in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e0f4a3bcd9c4b61adb59d6919901137 : SUCCESS in 10m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/add40013e53a4391a9322505672e697b : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b7c2f1750ab4be581d44d6d968b7230 : FAILURE in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/178bb914410c46639d3519b725629872 : SUCCESS in 13m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61f14da03dac4f208774c1f53caf9bef : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a828111d137c4e4ca3e24f21cca55df5 : FAILURE in 10m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08e5abc1aa224fa89399fe256c76145d : SUCCESS in 9m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/237ac9084ceb4ec89cc534a208c0d705 : SUCCESS in 1h 29m 01s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/c74e0054707645729d215458c29092c2 : RETRY_LIMIT in 5m 02s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c5ff2d6a27364c589353939d9d8c67c9 : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":7},{"id":"9422561588eeb171319a1709aa7c1cad2ef03f3c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-05 07:47:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e5310173443cf5ced7422c267ec9d964d958ba3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 09:21:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48d3b75c0cb24021be38ae8fad583a4e : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7b7fb8bba1144c2938c69756ff6fa3d : SUCCESS in 7m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d10cb75fead43dea866500a39059d45 : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce023099ad0e4f03a205186898ae0e3a : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678780216e6c49b190a2d37ab91e584d : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3596bb636084e2b9b84d97d72fa50d7 : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0796004555174939ad9e00a584b12897 : FAILURE in 4m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d9aed7ad03f46d0b017736c4b1cac51 : SUCCESS in 7m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/aea3a016634749c986398d56a89c90f2 : SUCCESS in 1h 28m 58s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/b11c7d20ae0a4114881e288ebc32a93e : RETRY_LIMIT in 3m 24s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b0b06eccfaa14e5196bb645bff23a5b9 : SUCCESS in 1h 20m 12s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":8},{"id":"dd09242cecafc1206908159e23b8a8cb2d330f00","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 06:43:01.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"c8bb474701451be6e5f238865e7a43d222f3098b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 08:21:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/650d0f4944c442f3844947165d4eef5a : SUCCESS in 10m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a53069a8edfd4ce49892ff1121f712f8 : SUCCESS in 11m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cdec626462a4819b3294ca9552a85a0 : SUCCESS in 9m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eea6d001f42640409231b393b746efe0 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f83fd714d3439eb9b04b02a9d1a676 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/774f32f5a0664989966dccf674516a00 : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a92f880b5254f2d94c56e46e03e2f23 : FAILURE in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ad81f4c119942bf85977b9dc12e7e9f : SUCCESS in 8m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a39d8f4514f04950be9d30d54255077e : SUCCESS in 1h 31m 09s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e29fd3a122fa4b39878279a3bb81fee5 : RETRY_LIMIT in 7m 38s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/d243139305cd4f5dbedf09760bb2f5ba : SUCCESS in 1h 21m 52s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":9},{"id":"a23f44f126ca495255cb033e1db3cf6f904c9b30","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-06 09:52:50.000000000","message":"Patch Set 9:\n\nMay be worth rebasing on https://review.opendev.org/#/c/703949 (easier to get test results and it should hopefully get merged soon anyway)","accounts_in_message":[],"_revision_number":9},{"id":"54d7ae1ac242bc9b266d41e72ed50d51d45bd04e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 11:53:29.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b07846e4871ededd02c04866dae85cc1dcb68c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 13:31:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24df31e7ba004e499b15c53a49a38f53 : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61f8d2d4cd9848259fe796966218db57 : SUCCESS in 7m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15f3d6d8db0342ad9ea7bcc794e4e57c : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60d2d14748bd4ee4bfe574da162e778a : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bedfa61c477f4dbda8db4228d8c0df43 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e46815ad04b64e36a88148a3dad578f8 : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/887ef45a251c402d9865a0655fb12021 : FAILURE in 6m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6022cd67015a4776ab0dab123905fbf6 : SUCCESS in 8m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7a858fea2c0c4e5a93340ee5dc6a56f6 : SUCCESS in 1h 28m 06s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/74ec851f886e4cb18c22a4f26247c275 : RETRY_LIMIT in 5m 07s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/050229a5c3124458945732b2649afdd0 : SUCCESS in 1h 17m 38s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":10},{"id":"3f8cf8fda50ad67eb5e79571fb20a98680f98a55","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 13:56:10.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"d85a0ef33a6756b916f2fc2fb37f2f3cb445b9e3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 14:51:25.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"a34ce7053febf961995b115a9babd555b1ea0c68","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 15:42:02.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"63720cad23138fbcbf65c59bd072172a43bb9d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 17:24:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02ff2c689b3b4007b4ec66c3987aaa89 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9f3241658c44d2d976d701971f69db8 : SUCCESS in 9m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd5a351ef8f44a928b2f7c4666e06203 : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f9f28201e4a48b79487456fa78f8785 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af0e3a9154c4e219f94da14b55b7587 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17e1b1d0bcb44b92a13a0154be028ef9 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c086ba599f0644d983997464b194a671 : FAILURE in 5m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53f6ab8649de45a8a213000149484508 : SUCCESS in 9m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/fab4dc5bff284ab08c163b4b7fd5adc3 : SUCCESS in 1h 35m 28s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/53636c3e828140688d4ff4cd29e17204 : RETRY_LIMIT in 6m 01s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4c985d3ebb3f4beb8316adaca9459102 : SUCCESS in 1h 19m 14s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":13},{"id":"7d2f1bf8f3c816c8109e58362495ba2c4d757d8e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 19:08:51.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"05bcfa91739400520d4faff91b6388d6c39e34f3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 19:09:06.000000000","message":"Topic changed from zuulv3 to bug/1859861","accounts_in_message":[],"_revision_number":14},{"id":"8815da08e96708e7e7780eaf776d5f0ece615c65","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 19:09:49.000000000","message":"Patch Set 14:\n\n\u003e May be worth rebasing on https://review.opendev.org/#/c/703949\n \u003e (easier to get test results and it should hopefully get merged soon\n \u003e anyway)\n\nGood tip, I spent the day with making ansible work to install frrouting things...","accounts_in_message":[],"_revision_number":14},{"id":"2c8b61d67c06b38a157417e42c3627e17d939312","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 19:20:25.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"94034bd02212dc7669490484a27b6c89c2e15dfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 20:43:37.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26e8719afdbd4085bd11ff4ad769aa75 : SUCCESS in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39e01cce5ab14c8eaa8b16602da1c8c7 : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32910bee13a44aee893bf8bdd5ecc5ea : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b671732c241245689447e9cdcf28d75a : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f342cb1a9e4ec8b5a2347b7924e9e8 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed8accdc9c8e4cc587d9ed6225cfa642 : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8070127eddc4a05bb0ae6e05df570f8 : FAILURE in 6m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d448b4564c144b38800cb4da479a7a3d : SUCCESS in 8m 24s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6c24b15b4b364bfb837757e060fee8d1 : SUCCESS in 1h 14m 13s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/07db03a2e01640cd9f1a652f1c51d880 : RETRY_LIMIT in 14m 48s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/413347ffb302461d8a645dfd795e78e9 : SUCCESS in 1h 21m 16s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":15},{"id":"c6122b6e8bcbcbebda86f11d559dbd4f7945cba9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-07 07:25:50.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"1326172efea51161ae439fefebe4c7b4e591f169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 09:08:28.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab0819610e74e4f93914b8a6f14bebb : SUCCESS in 12m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/850bdcf3372547708913ae087f42d9ee : SUCCESS in 8m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/699d57277f924836a3b2e7388842275d : SUCCESS in 12m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec538a369e754ce299a3e9ffd3935f0c : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9f7f3d41c04f8682089134388490b8 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be962ad2ffdb42a5ab5217f01cb81dab : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c83f921523034720a995b415e14e86c7 : FAILURE in 8m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe32cca02add420caa01c21758a1317a : SUCCESS in 8m 53s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6869bdecabdf4de5a0cf84258f2686ac : SUCCESS in 1h 34m 19s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/cfc2ec73fb054b85b46053c8a1e20515 : FAILURE in 25m 34s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c8c17770106341f3b37700d4a5c0e97d : SUCCESS in 1h 17m 03s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":16},{"id":"6150260bac498a1f2eb7976106f8a89651d4e26e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-27 10:36:17.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"852192bd0cd0e60acdd874ff007879b02a81f78d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 12:51:46.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/579e17c2bf1c4884a6fc038171a9d779 : SUCCESS in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19b49b3fb1e645129fd9f508fa8e2117 : SUCCESS in 11m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9328f7359eb24c18a0bbc4a4193fe0b3 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7345b710324adebdfb485dd6ec87a7 : SUCCESS in 15m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/827d51df36684da2ad5c6b7b77e0898a : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c640a18ea33449f0a66cf7bca3ba674d : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0dc4a3b5a84e4053ba572750a5b89a47 : SUCCESS in 8m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86579714acf347c2964204bca357b0cf : SUCCESS in 11m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2d0bdae63e354879991f8f358137baec : TIMED_OUT in 2h 10m 54s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/9a9673394060404ab169f6e2c52cd771 : FAILURE in 26m 25s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/7796c929807841928cc5b3298a2a08b0 : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":17},{"id":"ba92671a4d9323e1a4c16bc0d91c3a8845cbcd26","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-13 11:18:47.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"86dfaee5a0f1124a0aec13ed0b56ba1cc64c5740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 12:52:38.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86fb7c8021e14406b9bdf29163723cec : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1595eb678d5f4d8588049afa1db81a65 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bb7c14e3b8c401f9f55ad1295af5db0 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b183352acc534ebc831649183c9bc4dd : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2a90733ea824c5398fc2d9f28ede8f2 : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cd7e95a01024d9d9875d912aeeeaaf6 : FAILURE in 4m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a45b46cd0dd947b1a3ff40f6c838dc5e : SUCCESS in 8m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/04060e140930498b9a199ea365d56246 : FAILURE in 1h 25m 49s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/bea9b82847584238b4422e61f5dbe97c : FAILURE in 29m 58s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6c090b0c3706402897598ff1a5b08e4a : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":18},{"id":"91e03ad82d20da449dd56a2cba0ba5e953a4f5ec","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:53:15.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"da7dfc070aed65a2614fc822368641789da8d92a","revisions":{"e436d98f3391307fedfa4830b4820c4219d1bfed":{"kind":"REWORK","_number":1,"created":"2020-01-16 15:33:40.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/1"}}},"commit":{"parents":[{"commit":"46b433cac609b3fc9796217c6cb0ee1d86b80fca","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46b433cac609b3fc9796217c6cb0ee1d86b80fca"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:28:10.000000000","tz":60},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n\nCloses-Bug: #1859861\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e436d98f3391307fedfa4830b4820c4219d1bfed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e436d98f3391307fedfa4830b4820c4219d1bfed"}]},"branch":"refs/heads/master"},"f6e00cb113a0bd052005736438c91b0076fd8242":{"kind":"REWORK","_number":2,"created":"2020-01-17 15:38:54.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/2"}}},"commit":{"parents":[{"commit":"46b433cac609b3fc9796217c6cb0ee1d86b80fca","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46b433cac609b3fc9796217c6cb0ee1d86b80fca"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-17 15:36:41.000000000","tz":60},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) neutron fullstack NeutronConfigFixture assumed to have api_paste.ini\nin neutron/etc folder, which is not true if we execute bagpipe fullstack\nso I copied one to networking-bagpipe/etc folder.\n\nCloses-Bug: #1859861\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f6e00cb113a0bd052005736438c91b0076fd8242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f6e00cb113a0bd052005736438c91b0076fd8242"}]},"branch":"refs/heads/master"},"3cfd4b048d7c9bc653e1b727ee0d52972ae05482":{"kind":"REWORK","_number":3,"created":"2020-03-16 20:59:52.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/3"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-16 20:55:35.000000000","tz":60},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n\nTODO:\n1) stup mpls routing on zuul machines.\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3cfd4b048d7c9bc653e1b727ee0d52972ae05482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3cfd4b048d7c9bc653e1b727ee0d52972ae05482"}]},"branch":"refs/heads/master"},"b5c99a2be6007686023e392fd5be8f041010bbaf":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-03-16 21:00:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/4"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-03-16 21:00:50.000000000","tz":0},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n\nTODO:\n1) stup mpls routing on zuul machines.\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b5c99a2be6007686023e392fd5be8f041010bbaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b5c99a2be6007686023e392fd5be8f041010bbaf"}]},"branch":"refs/heads/master"},"c88da5f3593628ed8ffd6b6aa9f88228040aef35":{"kind":"REWORK","_number":5,"created":"2020-03-17 08:49:58.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/5"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-03-17 08:49:53.000000000","tz":0},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n\nTODO:\n1) stup mpls routing on zuul machines.\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c88da5f3593628ed8ffd6b6aa9f88228040aef35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c88da5f3593628ed8ffd6b6aa9f88228040aef35"}]},"branch":"refs/heads/master"},"ebee97a6d29c4223874e085248364e290be475e3":{"kind":"REWORK","_number":6,"created":"2020-04-01 21:44:56.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/95/702895/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/6"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-04-01 21:22:18.000000000","tz":120},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n\nTODO:\n1) stup mpls routing on zuul machines.\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ebee97a6d29c4223874e085248364e290be475e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ebee97a6d29c4223874e085248364e290be475e3"}]},"branch":"refs/heads/master"},"47c1df57f56a21d41ac2dc6c585d3b7e397fb92d":{"kind":"REWORK","_number":7,"created":"2020-04-03 07:58:20.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/7"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-03 07:57:33.000000000","tz":120},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/47c1df57f56a21d41ac2dc6c585d3b7e397fb92d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/47c1df57f56a21d41ac2dc6c585d3b7e397fb92d"}]},"branch":"refs/heads/master"},"f1e70949ddd021e8fbe34e0a3c58f85388479183":{"kind":"REWORK","_number":8,"created":"2020-04-05 07:47:13.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/8"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-05 07:46:41.000000000","tz":120},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1e70949ddd021e8fbe34e0a3c58f85388479183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1e70949ddd021e8fbe34e0a3c58f85388479183"}]},"branch":"refs/heads/master"},"e6c51a76eecbf0cb0e0dfc6c0155b4b5fa1b3b35":{"kind":"REWORK","_number":9,"created":"2020-04-06 06:43:01.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/9"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-06 06:42:58.000000000","tz":0},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e6c51a76eecbf0cb0e0dfc6c0155b4b5fa1b3b35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e6c51a76eecbf0cb0e0dfc6c0155b4b5fa1b3b35"}]},"branch":"refs/heads/master"},"69ff7bbae64e08f49d148119adb5fdef9ae95e18":{"kind":"REWORK","_number":10,"created":"2020-04-06 11:53:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/10"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-06 11:53:08.000000000","tz":120},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/69ff7bbae64e08f49d148119adb5fdef9ae95e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/69ff7bbae64e08f49d148119adb5fdef9ae95e18"}]},"branch":"refs/heads/master"},"e458176d60d5238b5ce8f29dfbe8368d54066009":{"kind":"REWORK","_number":11,"created":"2020-04-06 13:56:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/11"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-06 13:56:07.000000000","tz":0},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e458176d60d5238b5ce8f29dfbe8368d54066009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e458176d60d5238b5ce8f29dfbe8368d54066009"}]},"branch":"refs/heads/master"},"7a2bf6b60c27c6f7f9acd872f465781aaa864d05":{"kind":"REWORK","_number":12,"created":"2020-04-06 14:51:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/12"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-06 14:51:23.000000000","tz":0},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7a2bf6b60c27c6f7f9acd872f465781aaa864d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7a2bf6b60c27c6f7f9acd872f465781aaa864d05"}]},"branch":"refs/heads/master"},"7477c2322864c4e46f8b7930f729c8041aa13d3a":{"kind":"REWORK","_number":13,"created":"2020-04-06 15:42:02.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/13"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-06 15:41:23.000000000","tz":0},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7477c2322864c4e46f8b7930f729c8041aa13d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7477c2322864c4e46f8b7930f729c8041aa13d3a"}]},"branch":"refs/heads/master"},"955288379f08287af439448fd76e283e787282f7":{"kind":"REWORK","_number":14,"created":"2020-04-06 19:08:51.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/14"}}},"commit":{"parents":[{"commit":"adcb212b367b85cea81577583d8cf80e75852f8e","subject":"Switch fullstack job to Zuulv3 syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/adcb212b367b85cea81577583d8cf80e75852f8e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-06 19:07:23.000000000","tz":120},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/955288379f08287af439448fd76e283e787282f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/955288379f08287af439448fd76e283e787282f7"}]},"branch":"refs/heads/master"},"8b68410238fbb02cdd0d670199827430b07efd40":{"kind":"REWORK","_number":15,"created":"2020-04-06 19:20:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/15"}}},"commit":{"parents":[{"commit":"adcb212b367b85cea81577583d8cf80e75852f8e","subject":"Switch fullstack job to Zuulv3 syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/adcb212b367b85cea81577583d8cf80e75852f8e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-06 19:20:22.000000000","tz":0},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8b68410238fbb02cdd0d670199827430b07efd40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8b68410238fbb02cdd0d670199827430b07efd40"}]},"branch":"refs/heads/master"},"37df5bcc4bcebb686e9f37f1cf4b496661b19842":{"kind":"REWORK","_number":16,"created":"2020-04-07 07:25:50.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/16"}}},"commit":{"parents":[{"commit":"adcb212b367b85cea81577583d8cf80e75852f8e","subject":"Switch fullstack job to Zuulv3 syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/adcb212b367b85cea81577583d8cf80e75852f8e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-07 07:25:44.000000000","tz":0},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/37df5bcc4bcebb686e9f37f1cf4b496661b19842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/37df5bcc4bcebb686e9f37f1cf4b496661b19842"}]},"branch":"refs/heads/master"},"3a47e076e6cf0aafb658ef4ad98722a689680527":{"kind":"REWORK","_number":17,"created":"2020-05-27 10:36:17.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/17"}}},"commit":{"parents":[{"commit":"c5d52c16d26cd802c70e0fe983240ee406a2e256","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c5d52c16d26cd802c70e0fe983240ee406a2e256"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-05-27 10:34:22.000000000","tz":120},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3a47e076e6cf0aafb658ef4ad98722a689680527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3a47e076e6cf0aafb658ef4ad98722a689680527"}]},"branch":"refs/heads/master"},"da7dfc070aed65a2614fc822368641789da8d92a":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2020-10-13 11:18:47.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/95/702895/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/95/702895/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/95/702895/18"}}},"commit":{"parents":[{"commit":"7e96cc651394813c1dc80747186b6cfcaa173f14","subject":"Merge \"Add Python3 wallaby unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7e96cc651394813c1dc80747186b6cfcaa173f14"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-15 18:20:19.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-10-13 11:18:47.000000000","tz":0},"subject":"Fix fullstack tests","message":"Fix fullstack tests\n\nThis patch aimes to make fullstack tests working again, so it contains\nmultiple small fixes due to changes in neutron itself, and due to\npython2 - python3 differences for example.\n\nShort list of change reasons here:\n1) range() and filter() gives back an iterator from py3\n2) https://review.opendev.org/579100 removed central_external_bridge\nfrom neutron fullstack enviroment.\n3) configparser now expects config options to be strings.\n4) Change the way rootwrap is deployed, to make it effective for bagpipe\ntests.\n5) Make apipaste config work for bagpipe fullstack tests.\n6) setup mpls routing on zuul machines.\n\nTODO\n2) find out why ping is not working for bgpvpn connectivity tests\n\nCloses-Bug: #1859861\nDepends-On: https://review.opendev.org/712702\nChange-Id: I627217c439dd0c90361580956c9ca13480fc10fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/da7dfc070aed65a2614fc822368641789da8d92a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/da7dfc070aed65a2614fc822368641789da8d92a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
