)]}'
{"id":"openstack%2Fnetworking-bagpipe~702900","triplet_id":"openstack%2Fnetworking-bagpipe~master~I7838bd330a79635c528ec3ecdfd5825f8e39c578","project":"openstack/networking-bagpipe","branch":"master","topic":"lower-constraints","hashtags":[],"change_id":"I7838bd330a79635c528ec3ecdfd5825f8e39c578","subject":"Fix lower-constraints tox target","status":"MERGED","created":"2020-01-16 15:51:45.000000000","updated":"2020-04-09 10:24:39.000000000","submitted":"2020-04-09 10:21:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702900-1586427670784-b33c4e60","meta_rev_id":"42108b2eb12db3281dadecab020eee81e5aad1ce","_number":702900,"virtual_id_number":702900,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-04-09 09:53:47.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 10:21:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2020-04-09 08:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2020-04-08 10:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-04-09 01:13:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":1,"date":"2020-04-09 08:43:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 10:49:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-04-09 01:13:07.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-04-09 08:43:51.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-04-09 10:21:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3752ecdde4ae174efd07447fae843dc4b2cb8e31","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-16 15:51:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6c2d39af8559e3316ac176ed365e3e342cb650d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 18:00:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78cccd69469f4c929519c46d542e57a3 : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71dda2f99e8a46a5add2721fafaaa15a : SUCCESS in 9m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/086bb59011b74cc4909432b9767c666a : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c858c1ae770843e39b0ace0be740ec25 : SUCCESS in 8m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8f76fe489b924e1e82277dcf5d723959 : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e08735ccfea49259e26503db7263082 : FAILURE in 6m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f33bcd0b87214dd39f16559fa1cedae5 : SUCCESS in 11m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4eea8fffd3434790b1409368b84353fe : SUCCESS in 1h 38m 58s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e8967e28ccda465094a538d4d9fd76f1 : FAILURE in 23m 13s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0558bc0855ae404ba2c471ae01c6c93a : SUCCESS in 2h 00m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b5f343c939741c2575b2843591f30f80fde15055","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-17 09:21:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a31b6a369ac865a3dc78a1fb13fe419c31f73997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 12:15:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad56067faa734778a09a566c82452657 : SUCCESS in 13m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/308837a4c0334249a937cc53f7adf66e : SUCCESS in 10m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b98b957f0cc2471cb84715818614c095 : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880dd9f9f0394338a8c6b45cd0c49c0a : SUCCESS in 9m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/62dbf37a29684e00bfecb21ef408454a : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3878b6526dc94b3b8546f4e57c5a2b2c : FAILURE in 10m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/227cf6068f3a433d8fd07fb465f2698e : SUCCESS in 14m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b39146398856483795cc25e495b5d1dc : SUCCESS in 1h 30m 39s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ddb492f90e8c4af6852071f466860c73 : FAILURE in 25m 24s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/8a548929a91d4708a6411d377e59906a : SUCCESS in 1h 25m 16s","accounts_in_message":[],"_revision_number":2},{"id":"083bb6b359796d763f847d28e09e5e94829228ca","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-22 15:27:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3682544fb7c792ca9f4e5ad80e31031f67d39fb","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-01-22 16:02:38.000000000","message":"Patch Set 2:\n\nI would have expected that dependencies for neutron and neutron-lib would be picked from git rather than pypi, since these are declared in .zuul.yaml required-project for the openstack-tox-lower-constraints job.\n\nBut given the purpose of the lower-constraints job, this is probably intentional that they aren\u0027t.\n\nNow that I think about the topic, it seems likely to me that the job has been made non-voting for this particular reason.  It would be worth asking around to people knowing the check-requirements job if there is a solution to that.\n\nPerhaps one thing to try would be to formulate the dependency to neutron and neutron-lib in lower-constraints.txt as a git URL pointing to master branches.\n\nAnother option is: wait for the first beta release of neutron and neutron-lib to be pushed to pypi and then use this version in lower-constraints.txt (but keep the job non-voting because the same issue may arise again in the future).","accounts_in_message":[],"_revision_number":2},{"id":"d1678cee02aa3e8224fcd6242b0b369a39a3c2d2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-23 14:46:53.000000000","message":"Patch Set 2: Workflow-1\n\n\u003e I would have expected that dependencies for neutron and neutron-lib\n \u003e would be picked from git rather than pypi, since these are declared\n \u003e in .zuul.yaml required-project for the openstack-tox-lower-constraints\n \u003e job.\n \u003e \n \u003e But given the purpose of the lower-constraints job, this is\n \u003e probably intentional that they aren\u0027t.\n \u003e \n \u003e Now that I think about the topic, it seems likely to me that the\n \u003e job has been made non-voting for this particular reason.  It would\n \u003e be worth asking around to people knowing the check-requirements job\n \u003e if there is a solution to that.\n \u003e \n \u003e Perhaps one thing to try would be to formulate the dependency to\n \u003e neutron and neutron-lib in lower-constraints.txt as a git URL\n \u003e pointing to master branches.\n \u003e \n \u003e Another option is: wait for the first beta release of neutron and\n \u003e neutron-lib to be pushed to pypi and then use this version in\n \u003e lower-constraints.txt (but keep the job non-voting because the same\n \u003e issue may arise again in the future).\n\nYeah, perhaps the best is to wait for an ussuri neutron release.","accounts_in_message":[],"_revision_number":2},{"id":"593bf249e75031c0d9f9de78644a39714ba70bdd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 07:48:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"39f22ee2c8615caa0b5e13134a26f456601d56e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 09:36:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d8a260cda3146e9bf50ed36af61d32d : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dcb974760e845f8b7a3fe1b69efd3cc : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b15e8c86dce48bc8266304efe026cf4 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8e97403cef74518b36ef87fc8ea7987 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3d1914bed6b4216a963abb185029a28 : SUCCESS in 11m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f7db88982e894ed985f953bf94dca226 : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1294c3f98a6e4c45a8bab14abb73d0ae : FAILURE in 7m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa805227e90b425493afeada9a6efaa6 : SUCCESS in 8m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c0def9175de34a14a1dcae77b38605e0 : SUCCESS in 1h 25m 47s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/6d3a20b9a76d4b09b61b43fe6c9d17cf : FAILURE in 21m 42s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/05a9fcf46fb54d76828a9b3221d431cf : SUCCESS in 1h 44m 08s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":3},{"id":"37d4bf90c00d5821eae7d349c35f9eca9b78278d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 09:56:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5f8e65d08d59a9c86ccb9ecad44473ceb2484bb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 11:58:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdbdd67fdb604797921f73ed43215621 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fc81079f2a947c1aa9b0d26bf88d559 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f122163529be4988b7ac9c091fb98452 : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65b97d0a7bc04e9bb0abb4f9136190dc : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce93c62ecc244b98aba4ad30359d1258 : SUCCESS in 6m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4fda888b9a174a01a3d35f1ff1e8b97b : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c39da56027aa4738920261cd5883eb40 : FAILURE in 6m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bd2c446a4864b88a15992cb99c2d775 : SUCCESS in 9m 36s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ee8a5a69be414a1982284e954c48ea42 : SUCCESS in 1h 58m 50s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/7413b102a6db49929620e8eb131c042a : FAILURE in 23m 30s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/88b8a04fc2ad43659e5701190e42ab27 : SUCCESS in 1h 28m 14s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":4},{"id":"ecc199799a5523205de80c62f544368e7ee02c4a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 13:50:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e3a3ede71e263b29af726f78051c7e854c4dcfd4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 14:47:56.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"90b57168abf976363f95f1553e5ecebc5f45ce09","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-06 16:07:30.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"2afa92094649a7066b9d27fc0efe5d38c2fad8ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 17:44:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92f60895253c411bac257788d1f0abe1 : SUCCESS in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b4e9727cb8e432692688679d3a082cb : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0dbe8aae263f4731893a7af2d76dc200 : SUCCESS in 9m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa41d2dcc6174b48a8444fc2c4b2cf06 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8ed4eac5bc4a3387e7edf985678f2f : SUCCESS in 7m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aaee137b681b4fc48fb6802c6d12eb58 : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5744b4b35634f83ae712cbdc9397d3e : FAILURE in 5m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82168110544d48e39ce00aeb17e835ae : SUCCESS in 10m 21s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/65a05d62fbcb43a58e4f2a9a8a95dcda : SUCCESS in 1h 29m 38s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/51a70c5256dd49d4888a623af334ce38 : FAILURE in 20m 46s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/45ff3725f21b425f9454d544f1670988 : SUCCESS in 1h 24m 21s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":7},{"id":"6ee6589ae0cf40982537643ac7e83b502645c953","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-07 05:26:19.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"1f6a617361698a2e1a75f6dfe96a4aac871b1d30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 07:09:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4042842f024615829be1f18da188f9 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c47dde16d7674f6b9513bd0f8792e75a : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc39c97139ca41fb9398d4dedd6467f8 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2feed434060c43ccb2393057e6ce0fb8 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1e2bcdf2c74e75b1902f5c0be0ed90 : SUCCESS in 5m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3218e5fe38df420f9fbdbd3145ac0adb : SUCCESS in 2m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f9937632f2c4b11a8b93248f2ce7c84 : FAILURE in 4m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c7e4e6b5da147668af8f1548f820182 : SUCCESS in 7m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/77ab343345204bd2b3d7a020fca708ff : SUCCESS in 1h 35m 52s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f4f129f90df84dceba87926dd02b2d4b : FAILURE in 18m 42s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/bc4d21cf2ad64d1daf6a3ae75d31ddc0 : SUCCESS in 1h 19m 49s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":8},{"id":"ddb09e716fbd0527a00e38c95a973dba32aaeeef","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-07 07:47:05.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fb63d225e2ba3e519f6f0e0809a56abfca5d4e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 09:26:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c179c75344c4286b33790e9b1c0d6be : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88bddd6ff64245f3869255ffa9448446 : SUCCESS in 7m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0c2d3705fa94946a9114cf118ff636b : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7782dc3325f4bc39c637aabf8a55cb9 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e554e281ad449aea65b5e5a90e1fa3e : SUCCESS in 7m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d37a992fb8874c47a36a28236061918e : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82b92a55be03411da2f1639f120f00ea : FAILURE in 5m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2ea6c37593e4762ace691092b67b07d : SUCCESS in 7m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/83fd51494eb84a0191ea948815d4318a : SUCCESS in 1h 28m 00s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e8ad1443fc984c61bc2fb0ebaef16cdb : FAILURE in 18m 59s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/e702ec6c40794d2bb8d52b5f3685cfb0 : SUCCESS in 1h 31m 08s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":9},{"id":"04eff8bf2dcf9f5dd7dc89541765eb2d47da74d1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-07 12:13:51.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"09ec170828954590bd2d91258b903f5e89f97578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 13:44:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2ae09ee8f404946be55078ed842f518 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9877b445409424b96908db84a8b9162 : SUCCESS in 9m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30329e942b214404862aa796b47a95c3 : SUCCESS in 11m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32a69c87bfa84fb5a46f2e6923841690 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25299d5229bd442c9323d1d07f6358ad : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3f5977422f7451d86bca448c8a43c11 : SUCCESS in 5m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7bed00b384474c9c8f0cb17484d6eb38 : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b012144ece4f444fa71a293ad496eb9f : FAILURE in 8m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/895497c3e58548279dde4c67b1677bb0 : SUCCESS in 9m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f63e9a8925dd4c30994689e6956af8af : SUCCESS in 1h 21m 09s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ae93769ccf79476691758a1317ff7a4c : FAILURE in 21m 29s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/5d95f853b58d47e19a095b56e50c01a5 : SUCCESS in 1h 22m 12s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":10},{"id":"65da74e120581e38953b6c30e510729df6650d56","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-07 13:50:13.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"b1f732d275c9a47c42f2458a3d030abb9d035d1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 15:53:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b7dd5a24464ead82908946b18dc16b : SUCCESS in 11m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36a452ffcd8844c69a873d8a9249dee7 : SUCCESS in 11m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a725f704135146e18dc1269638a57705 : SUCCESS in 12m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b6509aa770b47c788474eca4851cc73 : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/157e6c4cfd96455c998a3b3c009d75fb : SUCCESS in 13m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8563331f5d7e425a9c6121b586ec636b : SUCCESS in 7m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f49d57279c2c47faa9ed3548ff377d4a : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fb7b8bf13814ee8a026d344e5ad9a91 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f49bfe66aa446beab4ac0806b88887f : SUCCESS in 9m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5af9132f8ca64291868e21e35a26aec2 : SUCCESS in 1h 39m 18s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/5ec69a31c5e24be18ea4ceaea2c00326 : FAILURE in 20m 04s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/015a9006d1b64e62b12ee938da3ba669 : SUCCESS in 1h 27m 26s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":10},{"id":"9cfe131effe966288ecf4f55975f1b2f9d2e5022","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-08 08:45:46.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"ce69c1715397075a7bf0d8da1e4944ee67debbe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 10:37:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16c2286d7b8e4111828b0b0b055dbb20 : SUCCESS in 11m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d992b2356bf4103a8295c4ba09ae972 : SUCCESS in 10m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f512073c4f5437e83dd7f6875e4192f : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85dd8f9ff8ce461382495c1d637b46bd : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3366221cdd3045a69fb9a0ce9359dce4 : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67831ca6336d4320b75511be2eb6a2d6 : SUCCESS in 6m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa3289fd8da6455daee52735bb4b52ed : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/877a81732930438599f503e69a0e5cc7 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df3cb7f315ad483c8cac2b7a6b28a82e : SUCCESS in 11m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7d6450f1f94d4d968e324b9ff9b6d096 : SUCCESS in 1h 28m 39s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2fb7a313c39b4c50b94b70708ca1d1fe : FAILURE in 24m 04s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/18bb79e3906f45ba8d5c106d7a6233f6 : SUCCESS in 1h 26m 55s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":11},{"id":"9994bfedddaba3908fd0ea11b3f80aa86e08767a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-08 10:49:43.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"b169e3dd37583cd2e975293b26a23476e2b7dfed","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-04-09 01:13:07.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"953767c955bc75d0509fe93bd469d82506d27e27","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-04-09 08:43:51.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"2d9ec005a558ca9007669595e4d915b894396944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 08:44:08.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"4462982858bd243acd8b90fcf1f9019264dd3760","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-09 09:53:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e3cd7b35d1bee295e46aae0e2c37b57729189f6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:21:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"e046083b613219c8e45ca8af8a4847d28dd4ef96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:21:10.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18bf84cadf74163ae9d8f49fb9bad1a : SUCCESS in 11m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa6c263d85b249a3a56d1ccd18bb7055 : SUCCESS in 9m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5a34fb9ebc34eba8e1807374c7c2d47 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5aa7ce3b0dc487298ef2e35e5785146 : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/119d61248f1044f482519d2d20a55cfe : SUCCESS in 7m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c9b8e457ccbe4554b1cc2578b1cc550d : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bab89737cf84beebb2bf76309cdfb41 : SUCCESS in 10m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/522289b1f8884ece94e17db135767234 : SUCCESS in 1h 19m 27s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/877cf8cbf8554b60933b6e40f4fe8f19 : SUCCESS in 1h 33m 11s","accounts_in_message":[],"_revision_number":11},{"id":"c151b69f8efdbb3f18e9ee7361560927cec69563","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:24:39.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc11a09d8dac49208d37cfc9a1702d54 : SUCCESS in 2m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c34661aef5cb47729af46ad27be6f933 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"2fd52987683dd602cee4bf356d8a882b948aea5e","revisions":{"b66d28f143de762992e25e22a51d4f1138743a92":{"kind":"REWORK","_number":1,"created":"2020-01-16 15:51:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/1"}}},"commit":{"parents":[{"commit":"46b433cac609b3fc9796217c6cb0ee1d86b80fca","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46b433cac609b3fc9796217c6cb0ee1d86b80fca"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:50:57.000000000","tz":60},"subject":"DNM: Fix lower-constraints tox target","message":"DNM: Fix lower-constraints tox target\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b66d28f143de762992e25e22a51d4f1138743a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b66d28f143de762992e25e22a51d4f1138743a92"}]},"branch":"refs/heads/master"},"1224d97b6667ec8b50a364240889f5e16daa5442":{"kind":"REWORK","_number":2,"created":"2020-01-17 09:21:41.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/2"}}},"commit":{"parents":[{"commit":"46b433cac609b3fc9796217c6cb0ee1d86b80fca","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46b433cac609b3fc9796217c6cb0ee1d86b80fca"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-17 09:21:09.000000000","tz":60},"subject":"DNM: Fix lower-constraints tox target","message":"DNM: Fix lower-constraints tox target\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1224d97b6667ec8b50a364240889f5e16daa5442"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1224d97b6667ec8b50a364240889f5e16daa5442"}]},"branch":"refs/heads/master"},"17acfd3dd2c08afa18193ea20baa594c49b8b0c2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-06 07:48:51.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/3"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-06 07:48:51.000000000","tz":0},"subject":"DNM: Fix lower-constraints tox target","message":"DNM: Fix lower-constraints tox target\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/17acfd3dd2c08afa18193ea20baa594c49b8b0c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/17acfd3dd2c08afa18193ea20baa594c49b8b0c2"}]},"branch":"refs/heads/master"},"ca0db98fe60a74309b55a70f624ddc505e640f7c":{"kind":"REWORK","_number":4,"created":"2020-04-06 09:56:47.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/4"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-06 09:56:29.000000000","tz":120},"subject":"DNM: Fix lower-constraints tox target","message":"DNM: Fix lower-constraints tox target\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ca0db98fe60a74309b55a70f624ddc505e640f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ca0db98fe60a74309b55a70f624ddc505e640f7c"}]},"branch":"refs/heads/master"},"477fd0e523b003cce34558f343567ab648901610":{"kind":"REWORK","_number":5,"created":"2020-04-06 13:50:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/5"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-06 13:43:39.000000000","tz":120},"subject":"DNM: Fix lower-constraints tox target","message":"DNM: Fix lower-constraints tox target\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/477fd0e523b003cce34558f343567ab648901610"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/477fd0e523b003cce34558f343567ab648901610"}]},"branch":"refs/heads/master"},"37f6175d6cd9895020b80cb97ba887b23540c73d":{"kind":"REWORK","_number":6,"created":"2020-04-06 14:47:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/6"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-06 14:47:53.000000000","tz":0},"subject":"DNM: Fix lower-constraints tox target","message":"DNM: Fix lower-constraints tox target\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/37f6175d6cd9895020b80cb97ba887b23540c73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/37f6175d6cd9895020b80cb97ba887b23540c73d"}]},"branch":"refs/heads/master"},"e3f6e39722b0cb0466ac62fe1bf3da1e1a8a70ea":{"kind":"REWORK","_number":7,"created":"2020-04-06 16:07:30.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/7"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-06 16:07:27.000000000","tz":0},"subject":"DNM: Fix lower-constraints tox target","message":"DNM: Fix lower-constraints tox target\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e3f6e39722b0cb0466ac62fe1bf3da1e1a8a70ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e3f6e39722b0cb0466ac62fe1bf3da1e1a8a70ea"}]},"branch":"refs/heads/master"},"f918ab7de36e2093851145261c5f029dc79c9ef7":{"kind":"REWORK","_number":8,"created":"2020-04-07 05:26:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/8"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-07 05:25:49.000000000","tz":0},"subject":"DNM: Fix lower-constraints tox target","message":"DNM: Fix lower-constraints tox target\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f918ab7de36e2093851145261c5f029dc79c9ef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f918ab7de36e2093851145261c5f029dc79c9ef7"}]},"branch":"refs/heads/master"},"11f918a1c11c66889d4d5982846752ce5dd25961":{"kind":"REWORK","_number":9,"created":"2020-04-07 07:47:05.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/9"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-07 07:46:44.000000000","tz":120},"subject":"DNM: Fix lower-constraints tox target","message":"DNM: Fix lower-constraints tox target\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/11f918a1c11c66889d4d5982846752ce5dd25961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/11f918a1c11c66889d4d5982846752ce5dd25961"}]},"branch":"refs/heads/master"},"5e4936aa9a16a5376f1b84f63f50b45828ec191d":{"kind":"REWORK","_number":10,"created":"2020-04-07 12:13:51.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/10"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-07 12:13:32.000000000","tz":120},"subject":"DNM: Fix lower-constraints tox target","message":"DNM: Fix lower-constraints tox target\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5e4936aa9a16a5376f1b84f63f50b45828ec191d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5e4936aa9a16a5376f1b84f63f50b45828ec191d"}]},"branch":"refs/heads/master"},"2fd52987683dd602cee4bf356d8a882b948aea5e":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2020-04-08 08:45:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/00/702900/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/00/702900/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/00/702900/11"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-01-16 15:44:01.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-08 08:45:43.000000000","tz":0},"subject":"Fix lower-constraints tox target","message":"Fix lower-constraints tox target\n\nSome requirements cleanup to use latest neutron-lib and neutron.\n\nChange-Id: I7838bd330a79635c528ec3ecdfd5825f8e39c578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2fd52987683dd602cee4bf356d8a882b948aea5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2fd52987683dd602cee4bf356d8a882b948aea5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
