)]}'
{"id":"openstack%2Fnetworking-bagpipe~707649","triplet_id":"openstack%2Fnetworking-bagpipe~master~Iefd0f0406f05439011fccac67302353c547ea33a","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/1853171","hashtags":[],"change_id":"Iefd0f0406f05439011fccac67302353c547ea33a","subject":"Remove ofctl code from bagpipe","status":"ABANDONED","created":"2020-02-13 14:26:47.000000000","updated":"2021-04-14 14:00:04.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b36d6f6a0801ccd78bf45df417ac189f38d83ffb","_number":707649,"virtual_id_number":707649,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"date":"2020-02-14 14:02:27.000000000","_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"date":"2020-03-05 19:39:13.000000000","_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-14 11:02:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":-1,"date":"2020-02-14 14:00:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 14:27:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-02-14 11:02:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-14 14:02:27.000000000","updated_by":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"reviewer":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"state":"REVIEWER"},{"updated":"2020-03-05 19:39:13.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2021-04-14 13:59:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"4ea8179a53afa61b6abbbf4a28b9df87fb1739a0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-13 14:26:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5540d98d4df1c36bed158c0f6eb31ccaaad2573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:37:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/664e933d1c1b434c92a48df896b2cd0e : FAILURE in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93c83e3e43af4cd0ba466dea50193fdb : SUCCESS in 8m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/916313128dfd46bbb877f0420168d132 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ebff11367b242788e75c7bd81189a84 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87845856a064154a9e03256b585f608 : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6b44dda3da74fcdadf03036547999fc : FAILURE in 7m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8860bb0d0c546c0b1fadb3772d69a91 : SUCCESS in 8m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7e1d184c25b5422e9fd2b8a1348e3047 : TIMED_OUT in 2h 05m 43s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/bf926a9c1bf64c60a518f69328a299ca : FAILURE in 18m 38s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6c7df534a3a04449b333f839d656b6bb : SUCCESS in 1h 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3fa52566538e185d6a2057b341a754bd242b3cc7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-14 08:56:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d731e3bcb1beaedaf35ab937a427a7b4a62110c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 11:02:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4fb9c6f84d84e7080586acdbdefa14a : SUCCESS in 11m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9acdbd291b94217b2446a17a275e208 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f04a6b3dd2234573b7c786b432db2b38 : SUCCESS in 12m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4a3e028a099432ebe12ff3cd002fff5 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb52795c74174afe8806dbfe16ce9471 : SUCCESS in 12m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8406a0b3c501436cbacdbaa6d8ded3f9 : FAILURE in 9m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/072428b511c5415ca276124949cd1a52 : SUCCESS in 12m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/23318796225d44c1a7baf9ded34b2437 : TIMED_OUT in 2h 04m 15s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2660bd1f7602433dae3161d8c0ba3980 : FAILURE in 26m 16s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4a80ac471427421e91d6c5d51b4fa189 : SUCCESS in 1h 28m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9a995424cf62ba28c271a8e483015311e98f0bd2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-14 13:34:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"284e7c5e0799110995e0c54fdb53bd41193a1263","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-14 14:00:11.000000000","message":"Patch Set 2: Workflow-1\n\nTrying to implement functional tests to cover this codepath","accounts_in_message":[],"_revision_number":2},{"id":"a167c9703bc748f8ed7a4979be6c8de6f5a08658","author":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"date":"2020-02-14 14:02:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aeb71754e1123777924884d0c3bc19008398bf97","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-17 16:11:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c1bdf18e027dd73c5edd73f7e4a8f953b3a16d4a","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-02-26 13:38:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10d4b0f2b53add8ec633109d98519f3aafd3d7c6","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-02-26 13:41:16.000000000","message":"Patch Set 2:\n\n+1 on intent here\n\nThat said I wonder I, beyond run_ofctl calls, there aren\u0027t places that implicitely rely on the ofctl code ; cases where support for some OVS fields or actions (for mpls) would exist in the ocftl code (because stuff is passed directly to the ovs-ofctl CLI) but would not have been added in the ofctl classes and/or os-ken. I\u0027m not sure this problem exist since I haven\u0027t followed the work around os-ken and OVS classes, but it\u0027s worh checking.  If such cases exist, fully removing the ofctl code will require first plugging these roles.","accounts_in_message":[],"_revision_number":2},{"id":"4ee66782471849797725dc7bec723ad0bd53340e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-28 14:08:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f47d09b8858a358c646c926b2582fdb7ee0a3c99","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-03-05 19:39:13.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b36d6f6a0801ccd78bf45df417ac189f38d83ffb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 14:00:04.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"63b35a3984de10449174f908b6a8d4e81aa3dd99","revisions":{"8b78ba683a802b673e2d6bc7940a484124e551d3":{"kind":"REWORK","_number":1,"created":"2020-02-13 14:26:47.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/49/707649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/49/707649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/49/707649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/49/707649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/49/707649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/49/707649/1"}}},"commit":{"parents":[{"commit":"f0c6cc49d585d21de8be882f4f2d5649cd833a07","subject":"Remove references for unittest2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f0c6cc49d585d21de8be882f4f2d5649cd833a07"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-10 15:27:16.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-13 14:26:20.000000000","tz":60},"subject":"Remove ofctl code from bagpipe","message":"Remove ofctl code from bagpipe\n\nChange-Id: Iefd0f0406f05439011fccac67302353c547ea33a\nRelated-Bug: #1853171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8b78ba683a802b673e2d6bc7940a484124e551d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8b78ba683a802b673e2d6bc7940a484124e551d3"}]},"branch":"refs/heads/master"},"63b35a3984de10449174f908b6a8d4e81aa3dd99":{"kind":"REWORK","_number":2,"created":"2020-02-14 08:56:51.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/49/707649/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/49/707649/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/49/707649/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/49/707649/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/49/707649/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/49/707649/2"}}},"commit":{"parents":[{"commit":"f0c6cc49d585d21de8be882f4f2d5649cd833a07","subject":"Remove references for unittest2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f0c6cc49d585d21de8be882f4f2d5649cd833a07"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-10 15:27:16.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-14 08:55:58.000000000","tz":60},"subject":"Remove ofctl code from bagpipe","message":"Remove ofctl code from bagpipe\n\nChange-Id: Iefd0f0406f05439011fccac67302353c547ea33a\nRelated-Bug: #1853171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/63b35a3984de10449174f908b6a8d4e81aa3dd99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/63b35a3984de10449174f908b6a8d4e81aa3dd99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
