)]}'
{"id":"openstack%2Fnetworking-bagpipe~708647","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ia4f2c78a11b6d53aee6eeb72c62ef78f885b9262","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/1863855","hashtags":[],"change_id":"Ia4f2c78a11b6d53aee6eeb72c62ef78f885b9262","subject":"Make releasenotes generation working again","status":"MERGED","created":"2020-02-19 13:15:25.000000000","updated":"2020-02-26 12:53:18.000000000","submitted":"2020-02-26 12:51:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708647-1582721499425-6db76d6a","meta_rev_id":"d15cbe33319d072436aa0fa55145858de9f1fde8","_number":708647,"virtual_id_number":708647,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-02-26 09:21:39.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 12:51:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2020-02-25 20:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2020-02-25 12:34:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-02-25 14:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-02-25 10:48:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":1,"date":"2020-02-25 20:42:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 13:18:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-02-25 10:48:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-25 12:34:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-02-25 14:41:02.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-02-25 20:42:39.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-02-26 12:51:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5256168cfe3fc8d44d56bebcd5984202a6d692d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-19 13:15:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f185dad39d54e051730db2e82ce0a3ced4f63384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 14:42:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb1c528432764fb49ae5193b9167ce6a : FAILURE in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc12cf6e152c44c3b499e40140bfbb28 : FAILURE in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3d2377582f34d39a31641854c629b46 : FAILURE in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf95dae5751a4c5ea90a5ee492fc02b3 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94fe3baf263b4885aa523c44fe7c1c8c : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/295431c1014841289cb48157e281870c : SUCCESS in 6m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1fc41c0a7ec34341864c9a7d4dffa4df : SUCCESS in 7m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01d99f7f3a894679b9382328176529dc : FAILURE in 10m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d4581ec1ec84762ada897f32c8efe85 : FAILURE in 6m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5e442acb613742fc926590026250a89e : FAILURE in 32m 23s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0d6dca8732c64365850c89cf915fd29d : FAILURE in 18m 54s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3d653128ed5740d3bf4a39e69cf99e90 : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"df1ccab03784cff7166384ca9eed7f8dbc52713c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-22 10:23:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nHello,\n\n-1 for constraints (see my inline comments)","accounts_in_message":[],"_revision_number":1},{"id":"093a1cc9230e9e9a3dec84536eb02d1402022570","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-24 11:51:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f86c92101acd98f87043afd897d3dabcc88eb4f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-24 11:54:08.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7a879e81e48f2954edc0ba66346c9a34cf996166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 12:19:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84a63faa546343eda03a4753536f2789 : FAILURE in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6737d4fe655d4adb92179278dd52ebd4 : FAILURE in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7429675a94904c57ac5df31197a1d702 : FAILURE in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/762eb6652ffb455a81bde871a44cf364 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68434d5fe9ee437298de2110d40d0fc6 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6291f22cffb43e6815de67ef92637ee : SUCCESS in 4m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4107bbcbe44c42219b550bb4bd00d600 : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a588a2f54aff4dd0aa7d3363e1d356f9 : FAILURE in 5m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec80c4c4dfbb4a739d0d6203b07547e9 : FAILURE in 5m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f4f5c9a8cb564f7098fa19f80f714ea6 : FAILURE in 12m 32s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ab0df0240a784533995ba143b94f6607 : FAILURE in 22m 28s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/475d277f95644ee497924868c98deeb6 : FAILURE in 14m 56s","accounts_in_message":[],"_revision_number":2},{"id":"fd0654a9733dfb0071a9d32a976bc31f370c45a2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-24 14:00:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"83d79d498cee406c1dff4bf842c27d45c26542cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:20:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09c6fb6efa1474d9abc76c9a7511a27 : FAILURE in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b535ef69895b454181eb183a1ebe5efc : FAILURE in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17890970f9aa487cbcf9db76e8696c67 : FAILURE in 9m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16d6581db7a6417597f6cd134e866967 : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4349a4717e4222a25f3f3baf242051 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b69146455c354fcabbd00186e41e1b05 : SUCCESS in 7m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5fa1d73cf0f7433ea3a621f58237f50e : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e2499e5d81c48878b09fcf89483d59c : FAILURE in 7m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c96aae41600b48b3ad57551ea001fbcf : FAILURE in 7m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4a6d4cadaffd4442b8abd727ced331df : FAILURE in 34m 34s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/4075bea91ff04b96a5b38e1246c1c4d5 : FAILURE in 20m 44s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/eeb4a157af484d66b5c0df4368811d6c : SUCCESS in 1h 16m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4c45238abee63482785d0e6978e80d01d6b7cad6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-25 06:00:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"cb39c77bb82912e7f6a17daa98ca60cdcc68b0f9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-25 06:01:20.000000000","message":"Topic changed from gate-failure-test to bug/1863855","accounts_in_message":[],"_revision_number":3},{"id":"271e59180fbcc52dcab22b2d5a6876256f710763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 07:44:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2ea8e774124281ae1e990a8b63aa35 : SUCCESS in 10m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86656e8a0ca3449f822e98b9404644cf : SUCCESS in 9m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c982ea0dc0d24b0fbf5e11de0093b0c5 : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edee5277c15b4bcba12f1583aff85b7f : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f71143bb6124adf9f564849d9c017f5 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b2167ad8a62460a80368f144d6a4e2b : SUCCESS in 7m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/95a1ddefe9764181b77a3053e4b75126 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0280286673804a81a22afb9cfa181b39 : FAILURE in 6m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f2df414e9ab40b8ba024d098f9e839c : SUCCESS in 9m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3591d05384db4610bdf8015104dc29cc : FAILURE in 33m 00s (non-voting)\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a6cb9b002c0644cdb7e98972b3b623ac : FAILURE in 21m 43s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b8cb1fa3ac7a4749b357a2f355cff107 : SUCCESS in 1h 32m 12s","accounts_in_message":[],"_revision_number":3},{"id":"4ca574c1a9cb12b82741849ee49f019c777db4bf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-25 07:53:30.000000000","message":"Patch Set 3: Code-Review+1\n\nhttps://9abde9dfbceba4b989ea-cd725815d30cf2ab8520aad695ae6c76.ssl.cf1.rackcdn.com/708647/3/check/build-openstack-releasenotes/1b2167a/docs/ looks OK","accounts_in_message":[],"_revision_number":3},{"id":"49abd2ad7db5a22bd53788b40b99c058d08ed42d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 08:41:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a0315f309b4b76a3e9d08f0f571a78cb0b55fff9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 08:47:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9aa09a21925697373da59a8bd0076a8a5506a487","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 09:08:12.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"62eeed019d73471031cef8f2e950cdf933379293","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-25 09:12:46.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"aff3fadcc3b5bacf3789f6a39c235247bbd6fe4c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-25 09:12:58.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"28a2a8e269d60c91e27c8e5c10c1ff21e464dac0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 09:16:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"76b5fa69b6861253bb0c34bad1616cdd70a64b42","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 09:17:14.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ec941654703dca3ab3de1829c6ea7abcd345fbb5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-25 10:48:49.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4d6393ef35526896790503886252657755c91d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:10:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1493b11ce7464bea98ab8cbe5472ffb1 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50ae5ace031a48b6863bf2f946b8b2a3 : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e73d5177217949249c79d88955179fc9 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e28f01d57b9548479dcfdb3530851059 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f620635abf4805bd75cd6e345282ee : SUCCESS in 12m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b70f387cb4c745fd945a7c06878fb276 : SUCCESS in 5m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3fb098f407c041e983dc0a6bf9233d9a : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63a7ebc15ec34dd789e2b39d2138f141 : FAILURE in 7m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8231843cefce438588430449d63dc1fa : SUCCESS in 8m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/dcd944a7497645009725ff5cdb20c1a7 : FAILURE in 37m 37s (non-voting)\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/cc5fb0733c894a6fa30b62a9899ce4d6 : FAILURE in 25m 08s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/39d1952d2b5e494fb639d15d44789da0 : SUCCESS in 1h 28m 57s","accounts_in_message":[],"_revision_number":5},{"id":"1f3ea4f6d65cbceccfa8406d1a0b8b749d6691cc","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 11:39:57.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"79aaa9454e9d49656a22f355e455c2b93dc2e23c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-25 12:34:04.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"b1789e1f2cec4aebb14cf6b94b77d255ff6c0d68","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 14:41:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ce7f77308572c45f8ade412053910a8f2558e6e7","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-02-25 20:42:39.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bd4ddc6188839ef782f1c6f214345bb99a0a8a41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 20:42:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0c4f367c5b8f9e5a9c35c49c9a16178485654895","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 22:03:29.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815a26be980041d58ca26a3b5b3dc8ec : SUCCESS in 10m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37c44e9c2d4644828d303428eb3f42dd : SUCCESS in 10m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/259c56886d3a450cbeceefba2e6bdcf8 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6a963c2a104414aa52079329fefe88 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1d8142d3f4a4a439036954d436e6606 : SUCCESS in 6m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/132c8507162543d3af876444dbca6bc6 : SUCCESS in 6m 10s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/28522bdac6314bc98e38f3ad99aebfa0 : FAILURE in 1h 18m 36s","accounts_in_message":[],"_revision_number":5},{"id":"068b66f7471743564f0645cfd2db5b87ffdb75fe","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-26 09:21:39.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"277143ccd541d89bff4375401a89589e90aa4a72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 11:14:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c759cb669de491fbd406efa4be38e4e : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c13eba49354540d4996cb048b00f7096 : SUCCESS in 8m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/109d552f64214cfbbad11bcfbb84f60d : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16029fb78ae2466f98fde03e8f302827 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380b580a15574e85a6041359a87c1768 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f36806f10d4408caf508557d8b962a7 : SUCCESS in 4m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e582c221ef9a4ff1880f586d4a9608d6 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d36f387d9b24dbe95531c15cab2b04e : FAILURE in 7m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4a982cb202e4fb6bdd9fbefeafa35d8 : SUCCESS in 10m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3d06ea7e12604f518db9dcba761b9a1b : FAILURE in 41m 08s (non-voting)\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/5460281eb41b495f83fe8946e6883d97 : FAILURE in 21m 41s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/02ec5a8b82ca4c48bdc0bfe168e11276 : SUCCESS in 1h 47m 44s","accounts_in_message":[],"_revision_number":5},{"id":"3826b8ee2f71b73f79d3bfc32593300a3e510eac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 11:15:01.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"95bc5cc77506c87fd7732bbadd8cc13fcbfefd4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:51:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3a89e575ed15614e2ece4f48fde45c2174808f4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:51:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23bc84ec52d147949a99c7861472bd89 : SUCCESS in 11m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77edeea61ce6418495909e96fec5a8eb : SUCCESS in 8m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c57f239b488d4671812c137eeb56660e : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e7abdc56e9c48a6b08beed890a0c720 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c19bb7f308404841a2962ee92f18ba41 : SUCCESS in 5m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a382b9237414247b2e9211c62d05668 : SUCCESS in 6m 39s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/73412c6d72634aafb11cf5a7e140617b : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":5},{"id":"d15cbe33319d072436aa0fa55145858de9f1fde8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:53:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c6933446f84993b8bb44604dd97119 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da28f90dded8442f8624f98d983b29ed : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cdba58cde05f0d5c108b27d11f79b5fcce53c924","revisions":{"a5e330a581c571655ddb2bd9017d2e2fed2e007d":{"kind":"REWORK","_number":1,"created":"2020-02-19 13:15:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/47/708647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/47/708647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/1"}}},"commit":{"parents":[{"commit":"f0c6cc49d585d21de8be882f4f2d5649cd833a07","subject":"Remove references for unittest2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f0c6cc49d585d21de8be882f4f2d5649cd833a07"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 13:12:52.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 13:13:54.000000000","tz":60},"subject":"Make releasenotes generation working again","message":"Make releasenotes generation working again\n\nChange-Id: Ia4f2c78a11b6d53aee6eeb72c62ef78f885b9262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a5e330a581c571655ddb2bd9017d2e2fed2e007d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a5e330a581c571655ddb2bd9017d2e2fed2e007d"}]},"branch":"refs/heads/master"},"53356a2340f9cc3ba523c3494caa80e57e0dd5aa":{"kind":"REWORK","_number":2,"created":"2020-02-24 11:51:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/47/708647/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/47/708647/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/2"}}},"commit":{"parents":[{"commit":"f0c6cc49d585d21de8be882f4f2d5649cd833a07","subject":"Remove references for unittest2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f0c6cc49d585d21de8be882f4f2d5649cd833a07"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 13:12:52.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-24 11:16:00.000000000","tz":60},"subject":"Make releasenotes generation working again","message":"Make releasenotes generation working again\n\nChange-Id: Ia4f2c78a11b6d53aee6eeb72c62ef78f885b9262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/53356a2340f9cc3ba523c3494caa80e57e0dd5aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/53356a2340f9cc3ba523c3494caa80e57e0dd5aa"}]},"branch":"refs/heads/master"},"8a4c2d5cbd1e80151bbb272fa1cda2360d591fec":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-25 06:00:54.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/47/708647/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/47/708647/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/3"}}},"commit":{"parents":[{"commit":"add225e7ead389d07cdc41ebd777460b0beba8da","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/add225e7ead389d07cdc41ebd777460b0beba8da"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 13:12:52.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-25 05:58:04.000000000","tz":60},"subject":"Make releasenotes generation working again","message":"Make releasenotes generation working again\n\nChange-Id: Ia4f2c78a11b6d53aee6eeb72c62ef78f885b9262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8a4c2d5cbd1e80151bbb272fa1cda2360d591fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8a4c2d5cbd1e80151bbb272fa1cda2360d591fec"}]},"branch":"refs/heads/master"},"a599a6e19af4022e245ab3db50833d6cc4377348":{"kind":"REWORK","_number":4,"created":"2020-02-25 08:47:41.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/47/708647/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/47/708647/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/4"}}},"commit":{"parents":[{"commit":"8c9f9143e158597f8765d77ba87e8bb25806b9b4","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8c9f9143e158597f8765d77ba87e8bb25806b9b4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 13:12:52.000000000","tz":60},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-25 08:44:19.000000000","tz":540},"subject":"Make releasenotes generation working again","message":"Make releasenotes generation working again\n\nChange-Id: Ia4f2c78a11b6d53aee6eeb72c62ef78f885b9262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a599a6e19af4022e245ab3db50833d6cc4377348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a599a6e19af4022e245ab3db50833d6cc4377348"}]},"branch":"refs/heads/master"},"cdba58cde05f0d5c108b27d11f79b5fcce53c924":{"kind":"REWORK","_number":5,"created":"2020-02-25 09:17:14.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/47/708647/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/47/708647/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/708647/5"}}},"commit":{"parents":[{"commit":"8c9f9143e158597f8765d77ba87e8bb25806b9b4","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8c9f9143e158597f8765d77ba87e8bb25806b9b4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 13:12:52.000000000","tz":60},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-25 09:16:52.000000000","tz":0},"subject":"Make releasenotes generation working again","message":"Make releasenotes generation working again\n\nChange-Id: Ia4f2c78a11b6d53aee6eeb72c62ef78f885b9262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/cdba58cde05f0d5c108b27d11f79b5fcce53c924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/cdba58cde05f0d5c108b27d11f79b5fcce53c924"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
