)]}'
{"id":"openstack%2Fnetworking-bagpipe~708648","triplet_id":"openstack%2Fnetworking-bagpipe~master~I423ed03ebb814154bb17462512732c39eb9f71c0","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/1863855","hashtags":[],"change_id":"I423ed03ebb814154bb17462512732c39eb9f71c0","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","status":"MERGED","created":"2020-02-19 13:15:25.000000000","updated":"2020-02-26 13:03:02.000000000","submitted":"2020-02-26 12:59:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708648-1582721972583-6731fd5b","meta_rev_id":"01bb1215e040c6a63f6420dc98506aa3de0c2481","_number":708648,"virtual_id_number":708648,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"date":"2020-02-26 08:41:40.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-02-25 13:14:36.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 12:59:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2020-02-25 13:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-02-25 14:41:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-02-25 14:41:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 13:18:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-02-25 10:50:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-25 13:40:16.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-02-25 14:41:17.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-02-26 08:41:40.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-02-26 12:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ceab2e5e18893439cfc2b7358e91767f0847a991","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-19 13:15:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36d12a6c15277d508de3060fd6f7d82289dd86c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 15:00:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9efee469a6cf4dbbb4992c6d658b951c : FAILURE in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86ae9cf8e29d4ee3bcf2e92295fff151 : FAILURE in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f05c8bfabc64d43bc1a3e02d981c9dd : FAILURE in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/467caa6e2fe842b294e9162c2dc11100 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fed0e7b05cb340419342baa009427ac1 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1521e68afd64fc9a51a5f409fbf8db2 : SUCCESS in 6m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/697790778b3d4b758e25c97c28e6f8b7 : FAILURE in 8m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d85aeb653924176a57ffecf59bdc0b7 : FAILURE in 7m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f413df8940f24117b61926f499b0bd67 : SUCCESS in 1h 36m 33s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/43c4221696e24c7a827cce15ea51e9a4 : FAILURE in 20m 08s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4d018fbbe25c4cb89b2f530498a61275 : SUCCESS in 1h 19m 22s","accounts_in_message":[],"_revision_number":1},{"id":"08ef2c953eb78e8c0538864d8bc0c0ffb155ecea","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-21 09:18:18.000000000","message":"Patch Set 1:\n\nJust a note for future myself, once this is in we can then switch neutron-tempest-plugin-bgpvpn-bagpipe back to voting in neutron-tempest-plugin","accounts_in_message":[],"_revision_number":1},{"id":"a2b854f9c0b44b6d449addfacd7ede7c16847271","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-22 09:59:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"82fa02f300c471e2591ca94c4e3a106a0485c267","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-22 10:19:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"819213632e926701e093b46171aeb5ac8fee767b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 11:21:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e559836a3744abb87775092c887efba : FAILURE in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6515e734b9934b2a8dad22df144f6bd6 : FAILURE in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73cf04e7fbd04729af879bbc0778f29e : FAILURE in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b9a4158a4ce40129f51c878554125bd : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7362161b62b345e8804cd79ece4df7e3 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13bc35fec70e4bbdb06aa0b1036e5142 : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/939e2a60be224f2fa067045f40df29ac : FAILURE in 6m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b30780f8678d49c48c728f13c8855b4d : FAILURE in 7m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3d4768b7d79d44438199ff57e386b166 : SUCCESS in 1h 21m 50s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/d4dede701d9243a283556ec24a2aaaf0 : FAILURE in 17m 07s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/01306ee87ea34e8e804eb37145920109 : SUCCESS in 1h 15m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3cfcd0dc200ed9acd741044a9df3c6a35e5d60f6","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-02-24 08:31:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87b7fb5320563397da51fe10af0b618eb9060f5e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-24 11:51:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4361e4daf4ec41c1f1c4c6024221e87dae1ab171","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-24 11:54:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a12bd6259ddd0dd7e4ffda26ff26e69037cdba5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 12:15:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08689ff7f19645a9a5487050616f92a2 : FAILURE in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5142fb6ac69a451f8804effbe25e077d : FAILURE in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b878f4ba3ba943c5b9fd7bbdafbc2f4f : FAILURE in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8b544597a6e47ab989dd9d8780ea8d4 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12975cea5cc046a8b6e3976556c20e8b : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e741b9aa1ec4854b376f062fd59aab2 : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8db52cccc5f4bba98c3196b916a8f49 : FAILURE in 9m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d27c7ffb199d49989cae9004cc946e2f : FAILURE in 6m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0c2ebf7f0e6c4b6983216fcd7b235bb9 : FAILURE in 18m 45s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/8f4128817dc943779ef4eb10a8e1dfba : FAILURE in 20m 56s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/cb893ef7d1a6493dbbd04630333f299c : FAILURE in 22m 27s","accounts_in_message":[],"_revision_number":2},{"id":"181fd680fe977e648cc119199e5cd0a91481f78b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-25 06:00:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f98777a5d67811874b60b4d865436004cfe7f996","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-25 06:01:40.000000000","message":"Topic changed from gate-failure-test to bug/1863855","accounts_in_message":[],"_revision_number":3},{"id":"eeb816eb1292346d846c178cd73e5f163168811b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 08:09:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70f3732feeb4bf0a18276740e36f1c3 : SUCCESS in 10m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4591c2f38f5c4d57bb115352f9659b86 : SUCCESS in 8m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e918971441d49ffa0d112bd4ec76293 : SUCCESS in 9m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/113579ae1aab498682b65a524dd4066d : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c063963a7e342e9bd4269407835c322 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c62639def404fd48878f2eae3095e79 : SUCCESS in 7m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bd6080f29e94828847c67ccaaebba92 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/862e08433f884a2e97c99d92d6115f11 : SUCCESS in 10m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f3a5d6eefee441298b49df1c57df46b0 : SUCCESS in 1h 21m 27s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/7289a381bbbc4a429d7dc99dd5aefffe : FAILURE in 20m 42s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b8388545cca448efaa529b5ef4fe316f : SUCCESS in 1h 23m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5ffc30c448a24956d6af66cbea7651f48247c679","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 08:47:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"643b46487b8ff592f376a2d6fbc289310173b69f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 09:11:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"196b77c933544d5d78f6a8d340fe4010eb221100","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-25 09:16:33.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"aa22c2882e9eba0819c3b700f6e8950e27dfc02a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-25 09:17:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"37357c387056b71e5c6f835be23eeab3bd3442be","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-25 10:50:33.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"343abc30dc4c781c0c144870b4f35e3e159aecf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:07:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a13f28b137d4298b5efdc147bfbf450 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/556b7f136fcf4c3980d9987274dbefc7 : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a98b6a22ecf54d839405cb39f8d4a59c : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6aee29da5676418abf84202f4f3f8bd0 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4897e59a8647ec819c4440de968e49 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74cf1cde5aa641fb8f293ae5de623eb0 : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83ac3d4fa8894f5d94883561e968ee49 : FAILURE in 6m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df3d5e94293e42e7bc666778be668017 : SUCCESS in 9m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a0568b42912848bea9ae7374a670e5c5 : SUCCESS in 1h 31m 37s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/339c6849555549058ba2320357a724d2 : FAILURE in 23m 23s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/5d322a22252f4a5cb91e382e5ad685d4 : SUCCESS in 1h 33m 13s","accounts_in_message":[],"_revision_number":5},{"id":"7fa2ddea6669bf1c9272a9085e6fb13b360a134f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 11:41:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5213fea7f5273a22fd0d68f490f828cd3cf38e5d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-25 12:56:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bcd2806b40aaea377e66d1f43977555c66a4836c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-25 13:14:36.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"904a21a8146a7e0c65059b6298845439c6ebcd20","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-02-25 13:40:16.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"6efaeabbd4d4508fd23efa83066dd9dc73afe6c4","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-25 14:41:17.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ec31400e6e64cefc01afbd7752cf4451a81befa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:48:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee23c7b39d1c4963b799a08dd0fa01f3 : SUCCESS in 10m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26cdc13479334670a181d8e45bb5c9a3 : SUCCESS in 10m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51a6d7d4b26847aaaee89d9d2780aed0 : SUCCESS in 10m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35af80636de743f9bf63377a1c4fef3e : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5afa29ae0bd34ea7b7d342faeacaa227 : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/727e9aef8b434f78836592c382ac73e7 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d709739d2ad48edb9495391c1b86b48 : FAILURE in 7m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aedd61f637074352b510f77131739a5a : SUCCESS in 10m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/90de680e610f434f8b5916e0225c2055 : SUCCESS in 1h 46m 17s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2223298a1fc44827a0eaa9af37fd160a : FAILURE in 27m 24s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/aa9262dc63d34f10b6c368a82c06772d : SUCCESS in 1h 30m 38s","accounts_in_message":[],"_revision_number":6},{"id":"098f5093981e75b47e4c93298b24c83e9683a5f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 20:42:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"afd5da5ac46488c9a3b5b892a6b8f61614bb7af3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 22:03:30.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"205d5d211191d2619c1ac58ef81c5a4179365498","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-26 08:41:40.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3d27ed3b6f3a289119ea1c13fd561f714995042a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:03:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48e67d19606a45a2a0217443086b9d35 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92afb93e66d94f1e80bf95de582b86fa : SUCCESS in 8m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f86833a68e943c38e09aad77d9474d2 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ed7b906b6c147eda60e6dc34ac0c325 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b7698ce51149809b47781b08818cb6 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c6630577720486fa8dfee2700574dbe : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6860976ec0344cbf9a948e9af2934314 : FAILURE in 6m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90479dabaee14a86a78ee981ac2537bf : SUCCESS in 9m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7198ad44aa67458d8715c3cf4b5db4f4 : SUCCESS in 1h 20m 13s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/27192366a3ee4418b57ba740cf9b98be : FAILURE in 19m 40s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/7926a008b4ee4fe69f60734d3bdf178e : SUCCESS in 1h 17m 58s","accounts_in_message":[],"_revision_number":6},{"id":"24bcf081a4e69a230882423f87509f031c7690d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 11:15:05.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0069234ad11fc4c294f7fc655b91dc84608f6056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:59:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8cbdfa2c842454086a559b46d66b29c : SUCCESS in 10m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18d59207c46c4914b423b428d71e7f97 : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58dca167b23e47c09270757add9f8fe5 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/443f1430d3384962ba54cc3db1fa767e : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/324c27d09dfe4fc2830339013de702dc : SUCCESS in 5m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/64ef41a9a3b347abac061dce920e4cca : SUCCESS in 1h 29m 49s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/59e9b92a40ba40ee9279506487b7e143 : SUCCESS in 1h 39m 21s","accounts_in_message":[],"_revision_number":6},{"id":"214356cfcca0839551b933b9c5b678ac8b1e7cec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:59:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"01bb1215e040c6a63f6420dc98506aa3de0c2481","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 13:03:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0338a0e79d4cf49aa658373a91a12d : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24ec81416199426282044bb38ba62a08 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"46777858a7a88630289a70b07365787ca9ec4f38","revisions":{"a8ac6703382394e3d942e211a8d121fe1fae1e02":{"kind":"REWORK","_number":1,"created":"2020-02-19 13:15:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/708648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/708648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/1"}}},"commit":{"parents":[{"commit":"a5e330a581c571655ddb2bd9017d2e2fed2e007d","subject":"Make releasenotes generation working again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a5e330a581c571655ddb2bd9017d2e2fed2e007d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 12:15:46.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 13:13:54.000000000","tz":60},"subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","message":"Remove bagpipe-bgp\u0027s deprecated action CLI option\n\nChange-Id: I423ed03ebb814154bb17462512732c39eb9f71c0\nCloses-Bug: #1863855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a8ac6703382394e3d942e211a8d121fe1fae1e02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a8ac6703382394e3d942e211a8d121fe1fae1e02"}]},"branch":"refs/heads/master"},"e723ce8f49a4e2bc341820eb34d8dbf2ef802fc1":{"kind":"REWORK","_number":2,"created":"2020-02-24 11:51:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/708648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/708648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/2"}}},"commit":{"parents":[{"commit":"53356a2340f9cc3ba523c3494caa80e57e0dd5aa","subject":"Make releasenotes generation working again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/53356a2340f9cc3ba523c3494caa80e57e0dd5aa"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 12:15:46.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-24 11:18:09.000000000","tz":60},"subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","message":"Remove bagpipe-bgp\u0027s deprecated action CLI option\n\nChange-Id: I423ed03ebb814154bb17462512732c39eb9f71c0\nCloses-Bug: #1863855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e723ce8f49a4e2bc341820eb34d8dbf2ef802fc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e723ce8f49a4e2bc341820eb34d8dbf2ef802fc1"}]},"branch":"refs/heads/master"},"5e64cbf266590001dcdfd73df16c36dad6112605":{"kind":"REWORK","_number":3,"created":"2020-02-25 06:00:54.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/708648/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/708648/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/3"}}},"commit":{"parents":[{"commit":"8a4c2d5cbd1e80151bbb272fa1cda2360d591fec","subject":"Make releasenotes generation working again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8a4c2d5cbd1e80151bbb272fa1cda2360d591fec"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 12:15:46.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-25 05:59:33.000000000","tz":60},"subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","message":"Remove bagpipe-bgp\u0027s deprecated action CLI option\n\nRemove old and deprecated CLI option for bagpipe-bgp, and make\nneutron-tempest-plugin-bgpvpn-bagpipe job voting again.\n\nChange-Id: I423ed03ebb814154bb17462512732c39eb9f71c0\nCloses-Bug: #1863855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5e64cbf266590001dcdfd73df16c36dad6112605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5e64cbf266590001dcdfd73df16c36dad6112605"}]},"branch":"refs/heads/master"},"be475e4e88e077300987864d4b505a535a6c2084":{"kind":"REWORK","_number":4,"created":"2020-02-25 08:47:41.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/48/708648/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/708648/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/4"}}},"commit":{"parents":[{"commit":"a599a6e19af4022e245ab3db50833d6cc4377348","subject":"Make releasenotes generation working again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a599a6e19af4022e245ab3db50833d6cc4377348"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 12:15:46.000000000","tz":60},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-25 08:46:35.000000000","tz":540},"subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","message":"Remove bagpipe-bgp\u0027s deprecated action CLI option\n\nRemove old and deprecated CLI option for bagpipe-bgp, and make\nneutron-tempest-plugin-bgpvpn-bagpipe job voting again.\n\nChange-Id: I423ed03ebb814154bb17462512732c39eb9f71c0\nCloses-Bug: #1863855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/be475e4e88e077300987864d4b505a535a6c2084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/be475e4e88e077300987864d4b505a535a6c2084"}]},"branch":"refs/heads/master"},"87e6629b334c90f1424a13dd10758936146923d6":{"kind":"REWORK","_number":5,"created":"2020-02-25 09:16:33.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/708648/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/708648/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/5"}}},"commit":{"parents":[{"commit":"a599a6e19af4022e245ab3db50833d6cc4377348","subject":"Make releasenotes generation working again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a599a6e19af4022e245ab3db50833d6cc4377348"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 12:15:46.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-02-25 09:16:28.000000000","tz":0},"subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","message":"Remove bagpipe-bgp\u0027s deprecated action CLI option\n\nRemove old and deprecated CLI option for bagpipe-bgp, and make\nneutron-tempest-plugin-bgpvpn-bagpipe job voting again.\n\nChange-Id: I423ed03ebb814154bb17462512732c39eb9f71c0\nCloses-Bug: #1863855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/87e6629b334c90f1424a13dd10758936146923d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/87e6629b334c90f1424a13dd10758936146923d6"}]},"branch":"refs/heads/master"},"46777858a7a88630289a70b07365787ca9ec4f38":{"kind":"REWORK","_number":6,"created":"2020-02-25 12:56:14.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/708648/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/708648/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/708648/6"}}},"commit":{"parents":[{"commit":"cdba58cde05f0d5c108b27d11f79b5fcce53c924","subject":"Make releasenotes generation working again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/cdba58cde05f0d5c108b27d11f79b5fcce53c924"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-19 12:15:46.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-02-25 12:53:47.000000000","tz":60},"subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","message":"Remove bagpipe-bgp\u0027s deprecated action CLI option\n\nRemove old and deprecated CLI option for bagpipe-bgp, and make\nneutron-tempest-plugin-bgpvpn-bagpipe job voting again.\n\nChange-Id: I423ed03ebb814154bb17462512732c39eb9f71c0\nCloses-Bug: #1863855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
