)]}'
{"id":"openstack%2Fnetworking-bagpipe~714837","triplet_id":"openstack%2Fnetworking-bagpipe~master~I1851fc2465bd20d73bedd795ca1cf6369af0995f","project":"openstack/networking-bagpipe","branch":"master","topic":"pcre_to_bindep","hashtags":[],"change_id":"I1851fc2465bd20d73bedd795ca1cf6369af0995f","subject":"Add pcre to bindep.txt","status":"ABANDONED","created":"2020-03-25 05:20:23.000000000","updated":"2020-04-01 07:37:59.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"77c4480bd7f23c27ffaec37750724605ab90ec06","_number":714837,"virtual_id_number":714837,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-27 08:51:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":-1,"date":"2020-03-27 07:18:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 05:40:37.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-03-25 11:44:34.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-03-27 07:18:44.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-03-27 08:51:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"738bc9772a81a7d223d6340d2dc148fcf06bb045","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-25 05:20:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f731c1188e61570b9ad166503fc35c41f3a24c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 06:42:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ad4f5b32204787a4c4a714f657e1bb : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/930669a8fe6d42f3adea2473d7914b6a : SUCCESS in 7m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d298f7f0bbf4bbcafd41f50b1a08238 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5531213c2c904b12b08110e8a1f66c60 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29eed6d53a9d46a7a8bd4efe32cfc811 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/342eafadb6d3442fac340970add6882a : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff3c57527edc4c2c8ab5e02accd3670e : FAILURE in 5m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78aa86563cfd42aea84fa4de9c5a1973 : SUCCESS in 7m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a844213a552d4345aff3306015098325 : SUCCESS in 1h 21m 19s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/4e4ec417b6114b21aa2746f479137aa8 : FAILURE in 15m 54s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ad1d441552344f169ac198c65296986a : SUCCESS in 1h 19m 00s","accounts_in_message":[],"_revision_number":1},{"id":"511340451f4be4055e798d80808afbd1b1ff76ea","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-25 08:59:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nThis patch seems necessary, since without it e.g. tox-py37 fails if pcre is not installed:\n\nscss/src/scanner.h:17:10: fatal error: pcre.h: No such file or directory [1]\n\nNevertheless it\u0027s a bit strange, that a horizon bindep needs to be added here. But I think it\u0027s fair as bindeps aren\u0027t handled recursively and for tests we need that in repositories where horizon is included.\n\n[1] https://zuul.opendev.org/t/openstack/build/514dc603d6a74fcea3c4939ecfc01655/log/job-output.txt#1432","accounts_in_message":[],"_revision_number":1},{"id":"5e540136fb0a01cdcf9033ddd5a369131bc24f07","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-25 09:15:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"890f5507efea1a95ab0a6c0501a2950c49eaf4b9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-25 09:17:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2d2a5de2ef6f9d1ee78c3e841d21e8b39deaa51d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-25 09:26:01.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks for the update, I think it\u0027s good like this.","accounts_in_message":[],"_revision_number":2},{"id":"71d950a510c9969b4e662425855880eeebc66996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 10:51:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f6743e6041485580caaca29c9cbe2f : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8453480e02114c3eac3be21222bceaf8 : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/481ba61dacf24b33a9341c753c3e2614 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59a5f77dcbee4f3ebf2987a674616d11 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c20f39b48464af183ffcff0c5a63f07 : FAILURE in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe7c6824712e451a9e3da5513a3837cf : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56aca017629f4b7ab370c16ffe53cca7 : FAILURE in 5m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d0c57d9f3d64c9a8f4559f3448c6f9a : SUCCESS in 9m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4668bd49297143f4b7e8ec2258902506 : SUCCESS in 1h 35m 14s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/053c7a25ccdb43f29d8bbb0c73a33d7c : FAILURE in 19m 51s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/35ac0fed3511459ba7319e293ea6a2dc : SUCCESS in 1h 22m 01s","accounts_in_message":[],"_revision_number":2},{"id":"2bb71f25d53398f1e4f0a79e5656d49341fd0d06","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-25 11:44:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f2cef5d1e36951f756010fbd20ac82a703001e6c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-26 10:31:03.000000000","message":"Patch Set 2:\n\nlibpcre is required by pyscss (one of horizon dependencies). pyscss is now working on make the C extension (libpcre-dev optional) again [1]. I believe it will land soon.\n\nI think it is good to have this till then.\n\nAnother solution is to release a beta version of networking-bgpvpn which moved horizon to optional requirements, but it is too much.\n\n[1] https://github.com/Kronuz/pyScss/pull/394","accounts_in_message":[],"_revision_number":2},{"id":"d0ef009dd4bebe1662fb99c1f563b1e191ae189c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-26 13:21:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03ecc6018502e6229d9234217b81d6129540cc8d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-26 13:22:01.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"522bd71209c55b54c401e68a985ae317d495c80c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-27 07:17:56.000000000","message":"Patch Set 2:\n\nI believe this is unnecessary now (as pyScss 1.3.7 is in upper-constraints).\n\nLet me use this to check if no bindep change works.","accounts_in_message":[],"_revision_number":2},{"id":"5f42f06278cc182bcce6d232f86ef996f0fcce8f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-27 07:18:26.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4c5533258b9e6672815733fa6ccf7c5e3b7ae77e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-27 07:18:44.000000000","message":"Patch Set 3: Workflow-1\n\nfake bindep change to trigger the gate.","accounts_in_message":[],"_revision_number":3},{"id":"44c0cc1d2043fb4c355f209bcdfa2e0b75b61a75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 08:51:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a551f4cd914ae08b4978d10df12c3e : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15f7b68e6c4c486fb9ac4da4b0da7108 : SUCCESS in 8m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8aaa3e436bce430dab9dab6e7f9618ca : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddc93b8e0db0431dbea292642993a6ef : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/821f76a032554ef09ab39c9950e2210a : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fec509c01d34421695d660ccec13fc48 : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6badbfa22c3e49a5b97a14f92a9a7b2b : FAILURE in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4966c47448e2478f80041357102c9c56 : SUCCESS in 8m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0be6d92db01640e4a9a180ec089f7553 : SUCCESS in 1h 25m 34s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a6fbf62035a74546a3ab61bfe23a429c : FAILURE in 20m 40s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/9538b6e472154a5b81a3dbbdf515d179 : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":3},{"id":"24a480eef7e1d2113d6b2162a7a739ada0ceaf80","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-04-01 05:36:03.000000000","message":"Patch Set 3:\n\nI believe this is no longer needed. All horizon issues around pyscss and setuptools 46 have been solved in the master branch.","accounts_in_message":[],"_revision_number":3},{"id":"c9d17caccc404891ab71c7ada46d4dd8211948bb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-01 07:37:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0c184dd0eaf3f46d90962ebb003a4b4707b30352","revisions":{"e5036df9129871aaeda9506dbbfc8c0f726c27d1":{"kind":"REWORK","_number":1,"created":"2020-03-25 05:20:23.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/37/714837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/37/714837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/1"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-25 05:17:57.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-25 05:17:57.000000000","tz":60},"subject":"Add pcre to bindep.txt","message":"Add pcre to bindep.txt\n\nAs networking-bagpipe depends on networking-bgpvpn and thus on Horizon\nin zuul, pcre is necessary to be installed for successfull gating.\n\nChange-Id: I1851fc2465bd20d73bedd795ca1cf6369af0995f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e5036df9129871aaeda9506dbbfc8c0f726c27d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e5036df9129871aaeda9506dbbfc8c0f726c27d1"}]},"branch":"refs/heads/master"},"a56c0a9508fa25d14d1f9ac8071d2228fe08147a":{"kind":"REWORK","_number":2,"created":"2020-03-25 09:15:24.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/37/714837/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/37/714837/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/2"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-25 05:17:57.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-25 09:11:27.000000000","tz":60},"subject":"Add pcre to bindep.txt","message":"Add pcre to bindep.txt\n\nAs networking-bagpipe depends on networking-bgpvpn and thus on Horizon\nin zuul, pcre is necessary to be installed for successfull gating.\n\nChange-Id: I1851fc2465bd20d73bedd795ca1cf6369af0995f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a56c0a9508fa25d14d1f9ac8071d2228fe08147a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a56c0a9508fa25d14d1f9ac8071d2228fe08147a"}]},"branch":"refs/heads/master"},"0c184dd0eaf3f46d90962ebb003a4b4707b30352":{"kind":"REWORK","_number":3,"created":"2020-03-27 07:18:26.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/37/714837/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/37/714837/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/37/714837/3"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-03-25 05:17:57.000000000","tz":60},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-03-27 07:18:14.000000000","tz":0},"subject":"Add pcre to bindep.txt","message":"Add pcre to bindep.txt\n\nAs networking-bagpipe depends on networking-bgpvpn and thus on Horizon\nin zuul, pcre is necessary to be installed for successfull gating.\n\nChange-Id: I1851fc2465bd20d73bedd795ca1cf6369af0995f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0c184dd0eaf3f46d90962ebb003a4b4707b30352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0c184dd0eaf3f46d90962ebb003a4b4707b30352"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
