)]}'
{"id":"openstack%2Fnetworking-bagpipe~716587","triplet_id":"openstack%2Fnetworking-bagpipe~master~I8f6ea2f2be093fc8f63d7eacf982e2bb990ff1b2","project":"openstack/networking-bagpipe","branch":"master","topic":"unittest.mock.functional","hashtags":[],"change_id":"I8f6ea2f2be093fc8f63d7eacf982e2bb990ff1b2","subject":"Remove the dependency on the \"mock\" package","status":"MERGED","created":"2020-04-01 13:10:45.000000000","updated":"2020-05-10 16:54:02.000000000","submitted":"2020-05-10 16:52:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716587-1589129565020-eaa21e4c","meta_rev_id":"6946c44f148e696c2dcd873a067e20722ba04a04","_number":716587,"virtual_id_number":716587,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-10 16:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-06 19:14:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2020-05-05 14:51:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-05-05 15:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-10 15:23:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 11:55:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-05 14:51:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-05-05 15:47:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-05-10 15:23:37.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-05-10 16:52:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a86c36463e11bf90971969f988ae8d85aa2e4354","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-01 13:10:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eda0fd48df61a9e8f401d3c596020d4c403c7436","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 15:16:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1f476946ba4e1e842692fdbb89ceac : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d8009c23a5341e9938ae9414a660d80 : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46ede8dd6aeb4a30b584683be3e6a9e5 : SUCCESS in 9m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c01fc401ec7a489fa76fa328e56f9955 : FAILURE in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7134715d0215499680d64592dfd5234f : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cb9ea55c4084dedbe889ef79ff9929b : FAILURE in 5m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edc51baf4b94472bbe32f1d8a6661a78 : SUCCESS in 8m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/af64b3511e7e41e1b517735f327532af : SUCCESS in 1h 29m 26s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/fc9f70208a0c4bfdb8060f76a04360dc : FAILURE in 23m 52s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3d18cfc669b7457180f7944022701423 : SUCCESS in 1h 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"264fcbf7568445de8b607f305f81b2b3c5c58dba","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-02 22:41:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b2f40654b85dbf44cf6adf97c211db98591485f4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-02 23:16:35.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks OK, py38 failure is because of mock in neutron-lib","accounts_in_message":[],"_revision_number":1},{"id":"b0b1667f14974154e9231bcb7a0f9351e21e63e2","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-04-09 08:48:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI understand that https://review.opendev.org/702900 will sort out the py3.8 issue, so let\u0027s go with this one.","accounts_in_message":[],"_revision_number":1},{"id":"eeae53c61c6464b151b2d12f40d4fd9de9d854aa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-20 11:55:08.000000000","message":"Patch Set 1:\n\nNeeds to remove from requirements too","accounts_in_message":[],"_revision_number":1},{"id":"1f3e2c3cedffaf65fc22d224eb50e4180fecdf1e","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-04-21 19:42:36.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"29bc99a882570b8579d1f0d424d95a809616052f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-24 07:43:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e20446e8983304ae6059d32145ed36a11003a4e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 07:44:07.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"94bf65d49ed84e7f8789766a89ec3ba7b0b67aed","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-24 07:49:37.000000000","message":"Patch Set 2:\n\n\u003e Needs to remove from requirements too\n\nThanks, I removed from lower-constraints.\n\nAfter rebase, I can\u0027t see where\u0027s the conflict....","accounts_in_message":[],"_revision_number":2},{"id":"c45417bd01631c180c7a4756e480d9b3aeac11a9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-24 13:11:26.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b6d7ff93a158720e680edc3f53024ab6ad60ade8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 13:11:41.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"0dfbba025354f88908d8ad4feb4e98ab15505a09","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-05 12:40:59.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5dd92bd09247e475c444a73848d5949c254f3ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 14:33:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c30fee8b311e4192bcb2773d4b5a2010 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9b6087ce4d64043b6b6a61a1528b8b4 : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25ebf492bf9f42659d406cc4df802eeb : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b850e57acd14dff9933fd9b4c4b42ba : SUCCESS in 6m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/be65882509cf4d318e89837f540336be : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8af3978e42f544b4a5e70b947b0e1041 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41cd0486881b46a58869543fec13de75 : SUCCESS in 9m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f799fa1e31e841f292fa386ad41fefb1 : SUCCESS in 8m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/41f96b11b5804390954f8dd13dd19ef2 : SUCCESS in 1h 25m 39s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/9a10567fcc1949329423952d37a527f2 : FAILURE in 15m 55s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/988e2b8b2f694166b3199f5fe1a7a293 : SUCCESS in 1h 45m 06s","accounts_in_message":[],"_revision_number":4},{"id":"845fffd677d105ccd3c374f69acf38f7078d51f5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-05 14:51:37.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5bdcdf2ce4ab58de81b7cd0bc64c5de2c58b0823","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-05 15:47:04.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"439fabcedebfce2864b909eab3eebfcd52743809","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-05-06 19:14:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0e5d5aa893fc25100a5cf1db0ed741332a462102","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-05-10 15:23:31.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"ea23bb65e8104564129d7214bc7a2ffb300045a1","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-05-10 15:23:37.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"610423d0921175c27186d040ef2e233cdcfaae49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 15:23:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"120886064ffd3d7724c8a0fd209b42e1e4d0a27b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 16:52:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed472731e06440b89d69f9dd0563556 : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21a939ab42534351958808366cc277eb : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49ebc7ae27cf4ffea4f4f7a393d2e4b7 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8acb1a2bef4a1ea1ae00ba8dcacfee : SUCCESS in 7m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3880e391e534312b9f06ce9508f0cd2 : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2875f507c3394d90a4af65e763c9bdd5 : SUCCESS in 8m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/13888d6db138495f8b73af83776434e5 : SUCCESS in 1h 26m 42s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c06f9d4e27824abb8e599f3773b7e89a : SUCCESS in 1h 27m 56s","accounts_in_message":[],"_revision_number":4},{"id":"4a3acd534eca4deeacb668ba32752d5654b87483","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 16:52:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8d615af6c0089b51340eb38aabaa22ab1bc9c7cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 16:54:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c8742c71a746cc97b41cdb76668811 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d3d94117dec56073136898924a72798019eb2ac2","revisions":{"1e047a182b0ca0f52584f12c24bffbefd2ce063b":{"kind":"REWORK","_number":1,"created":"2020-04-01 13:10:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/716587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/87/716587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/1"}}},"commit":{"parents":[{"commit":"46777858a7a88630289a70b07365787ca9ec4f38","subject":"Remove bagpipe-bgp\u0027s deprecated action CLI option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/46777858a7a88630289a70b07365787ca9ec4f38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-01 11:36:24.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-01 13:09:32.000000000","tz":120},"subject":"Remove the dependency on the \"mock\" package","message":"Remove the dependency on the \"mock\" package\n\nDepends-On: https://review.opendev.org/716049\nDepends-On: https://review.opendev.org/716584\nChange-Id: I8f6ea2f2be093fc8f63d7eacf982e2bb990ff1b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1e047a182b0ca0f52584f12c24bffbefd2ce063b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1e047a182b0ca0f52584f12c24bffbefd2ce063b"}]},"branch":"refs/heads/master"},"4b3a21c1a87a1a42981b819c9f890687331355d8":{"kind":"REWORK","_number":2,"created":"2020-04-24 07:43:49.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/716587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/87/716587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/2"}}},"commit":{"parents":[{"commit":"dc7bafef95a43c47f72dcd69d0b6238919d3a42e","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dc7bafef95a43c47f72dcd69d0b6238919d3a42e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-01 11:36:24.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-24 07:43:14.000000000","tz":120},"subject":"Remove the dependency on the \"mock\" package","message":"Remove the dependency on the \"mock\" package\n\nDepends-On: https://review.opendev.org/716049\nDepends-On: https://review.opendev.org/716584\nChange-Id: I8f6ea2f2be093fc8f63d7eacf982e2bb990ff1b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4b3a21c1a87a1a42981b819c9f890687331355d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4b3a21c1a87a1a42981b819c9f890687331355d8"}]},"branch":"refs/heads/master"},"11d43f53566eeffa5cc2436afc91bbd364608632":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-24 13:11:26.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/716587/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/87/716587/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/3"}}},"commit":{"parents":[{"commit":"dc7bafef95a43c47f72dcd69d0b6238919d3a42e","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dc7bafef95a43c47f72dcd69d0b6238919d3a42e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-01 11:36:24.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-24 13:10:53.000000000","tz":120},"subject":"Remove the dependency on the \"mock\" package","message":"Remove the dependency on the \"mock\" package\n\nDepends-On: https://review.opendev.org/716584\nDepends-On: https://review.opendev.org/716049\n\nChange-Id: I8f6ea2f2be093fc8f63d7eacf982e2bb990ff1b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/11d43f53566eeffa5cc2436afc91bbd364608632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/11d43f53566eeffa5cc2436afc91bbd364608632"}]},"branch":"refs/heads/master"},"d3d94117dec56073136898924a72798019eb2ac2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-05 12:40:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/716587/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/87/716587/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/87/716587/4"}}},"commit":{"parents":[{"commit":"7190eb584495307b8f000776a8bb1d4637bc386c","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7190eb584495307b8f000776a8bb1d4637bc386c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-01 11:36:24.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-05 12:40:59.000000000","tz":0},"subject":"Remove the dependency on the \"mock\" package","message":"Remove the dependency on the \"mock\" package\n\nDepends-On: https://review.opendev.org/716584\nDepends-On: https://review.opendev.org/716049\n\nChange-Id: I8f6ea2f2be093fc8f63d7eacf982e2bb990ff1b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d3d94117dec56073136898924a72798019eb2ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d3d94117dec56073136898924a72798019eb2ac2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
