)]}'
{"id":"openstack%2Fnetworking-bagpipe~725361","triplet_id":"openstack%2Fnetworking-bagpipe~master~I5ea4bb27361df3c489deedb51d7ca8ea64bb923b","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/1863021","hashtags":[],"change_id":"I5ea4bb27361df3c489deedb51d7ca8ea64bb923b","subject":"Monkey patch original current_thread _active","status":"MERGED","created":"2020-05-04 18:19:07.000000000","updated":"2020-05-07 09:42:47.000000000","submitted":"2020-05-07 09:18:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725361-1588843134483-0280b767","meta_rev_id":"928a12f2cdb576baed4b67e688a3051e585a57df","_number":725361,"virtual_id_number":725361,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 09:18:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-05-07 07:25:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-05-07 07:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-05-07 07:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-05-07 07:35:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 07:25:32.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-05-07 07:29:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-07 07:35:54.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-05-07 09:18:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6edab4490332752e8c580c799a85ef7aabe8aa46","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-05-04 18:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb5d109b8cc84fa1c56e1ee2566d4c2bffa6f33f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 19:48:18.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e0e0bf863e14e15b18ca2714ff5ff24 : FAILURE in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/972db3b5065f4eba960020e545fad50b : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/429e31be2b22483dac58bd42fcf41a23 : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db060d87d3e14dad83b1e73c17d5241d : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4af17cb69d94e46827aed14d7b58fad : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec9397b1538f44459ad9594e0023ba07 : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05beb1c323314a76a33a13920a4f454c : SUCCESS in 9m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4136a623908d48078e4abbb4b06d5c1d : SUCCESS in 1h 27m 09s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/634d5b71294044759bcf809b69ff91b8 : FAILURE in 20m 11s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6a0aa49dc67c43e1b1f904786cab074b : SUCCESS in 1h 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"503dc9f71ace0a10584b7a5df1b31627d7b793b9","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-05-04 20:39:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b5edd9e29ceadfd67cb87f8d849b9fd47d2cb48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 22:25:14.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255390ecd3f94538af85d59fe70fc75e : FAILURE in 10m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cae679695c1e4dc99947567332189aa6 : SUCCESS in 12m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05dfc12ae6cf4442accf6a395425e299 : SUCCESS in 12m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3a605c98f6454ab3d9e45630b16543 : SUCCESS in 8m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be169672f52547be9e60d8f377d57ad6 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb89744b0f7c4f8a8b9fabae29735149 : SUCCESS in 13m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c76280a103144e10842143e36f8705ef : SUCCESS in 10m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b67d110242b64e7b80587f55d38af319 : SUCCESS in 1h 43m 00s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/d0b3df8671254af48cdfcd9650be695d : FAILURE in 15m 36s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/26cb1f4360b94fcfa4522399560a20a0 : SUCCESS in 1h 25m 04s","accounts_in_message":[],"_revision_number":2},{"id":"046747941cf8133bb789ba3e19138f935ab0a35b","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-05-06 16:50:06.000000000","message":"Patch Set 2:\n\nre: pep8: E0401: Unable to import \u0027__original_module_threading\u0027 (import-error) -- I think this can safely be ignored. Once the eventlet.patcher.original() function gets called, sys.modules[\u0027__original_module_threading\u0027] \u003d real_mod will be defined.","accounts_in_message":[],"_revision_number":2},{"id":"88ebe13eae34946c0e58d1eb3bd40ad07c159989","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-05-06 16:50:39.000000000","message":"Patch Set 2:\n\nMore specifically, eventlet.patcher.original(\u0027threading\u0027)","accounts_in_message":[],"_revision_number":2},{"id":"99c76d72f3368f0c713fc2913ddcd51d37cd6d53","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-05-06 17:05:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4e6405a40ba4228ebd5db420d6e37178cf106844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 18:45:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dc8c142c76846f8bd915d17c219359b : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79c5812bf7044c97b5d4408b92800967 : SUCCESS in 12m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b678a30f434e473a90940763aa1a4e23 : SUCCESS in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8934cc5dd8bd44fd9d4bb0a7f138ef45 : SUCCESS in 17m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01eff16da96f4dfbb5060bacb470562f : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/779839b530dc424386ba625dc76863cc : SUCCESS in 12m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7725bfcbee7d42dd84c855ce0675a007 : SUCCESS in 10m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/55ad2d8d9e1e4cfeadde8f62f3a27a06 : SUCCESS in 1h 37m 26s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/c78533c7f38b446fb8797158eef8c4a7 : FAILURE in 26m 21s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/59b0a35633994f4b818698a39a064f1a : SUCCESS in 1h 27m 34s","accounts_in_message":[],"_revision_number":3},{"id":"d3054e96caa2cf918e70e10b90118a80adc926e1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-07 07:25:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"15acb11b55dbfe8a43c369e694db6480f21d09c3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-07 07:29:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c5853d7c08c934b4dc65e2ab7706d61f4a89790e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-07 07:35:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7f6c5c8f7e54d81e691a9b3b04c187c1d8238ff6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 07:36:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a037fbadcaa516a1b33c35b1fc860dda89260648","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 09:18:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/424af0c112a04d7b870783bf3e939053 : SUCCESS in 9m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4d3741d50a84dd2b35702ea590e8b4d : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8594cdd7ea1342dda557af4712439517 : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/106e043e78e844de98ff40c26c79e64a : SUCCESS in 9m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d40c83b54b314848b4ee546f3925c741 : SUCCESS in 10m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/715215e71e0f40f88800f3e4645db892 : SUCCESS in 1h 40m 50s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/51e65df43b6a462b8ecdb2000df79295 : SUCCESS in 1h 16m 42s","accounts_in_message":[],"_revision_number":3},{"id":"197c981f937db94478042a7c3af34e5796261339","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 09:18:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c1007eea72e90ef27576b83bec463bbb27e72de2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 09:21:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0932c9e2a067496abefb9db1dac998f6 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":3},{"id":"a3d16158eb57bad957b663d4b7f4c50f9243ca1b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-07 09:42:35.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 7b98c0f2312f1afadf5fd94ae387baf530b4a402","accounts_in_message":[],"_revision_number":3},{"id":"9b1de80f166b2d57c5043ad8ea95922bb34a7c61","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-07 09:42:47.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit bb21621c2e4a25d9d38f719b5ac83160e830e2ef","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f0db421843f1161f4daa1e08b83eb2b11e78d9dd","revisions":{"a321920de1c5c1451760a8a6775c679bcca84a9d":{"kind":"REWORK","_number":1,"created":"2020-05-04 18:19:07.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/61/725361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/61/725361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/1"}}},"commit":{"parents":[{"commit":"7190eb584495307b8f000776a8bb1d4637bc386c","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7190eb584495307b8f000776a8bb1d4637bc386c"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 18:18:49.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 18:19:06.000000000","tz":-240},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I5ea4bb27361df3c489deedb51d7ca8ea64bb923b\nCloses-Bug: #1863021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a321920de1c5c1451760a8a6775c679bcca84a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a321920de1c5c1451760a8a6775c679bcca84a9d"}]},"branch":"refs/heads/master"},"148a372f2ab5513e917be2d8c2d27a36aa233952":{"kind":"REWORK","_number":2,"created":"2020-05-04 20:39:50.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/61/725361/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/61/725361/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/2"}}},"commit":{"parents":[{"commit":"7190eb584495307b8f000776a8bb1d4637bc386c","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7190eb584495307b8f000776a8bb1d4637bc386c"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 18:18:49.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 20:39:46.000000000","tz":-240},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I5ea4bb27361df3c489deedb51d7ca8ea64bb923b\nCloses-Bug: #1863021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/148a372f2ab5513e917be2d8c2d27a36aa233952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/148a372f2ab5513e917be2d8c2d27a36aa233952"}]},"branch":"refs/heads/master"},"f0db421843f1161f4daa1e08b83eb2b11e78d9dd":{"kind":"REWORK","_number":3,"created":"2020-05-06 17:05:45.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/61/725361/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/61/725361/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/61/725361/3"}}},"commit":{"parents":[{"commit":"7190eb584495307b8f000776a8bb1d4637bc386c","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7190eb584495307b8f000776a8bb1d4637bc386c"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 18:18:49.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-06 17:05:39.000000000","tz":-240},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I5ea4bb27361df3c489deedb51d7ca8ea64bb923b\nCloses-Bug: #1863021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f0db421843f1161f4daa1e08b83eb2b11e78d9dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f0db421843f1161f4daa1e08b83eb2b11e78d9dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
