)]}'
{"id":"openstack%2Fnetworking-bagpipe~726073","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fussuri~I5ea4bb27361df3c489deedb51d7ca8ea64bb923b","project":"openstack/networking-bagpipe","branch":"stable/ussuri","topic":"bug/1863021-stable/ussuri","hashtags":[],"change_id":"I5ea4bb27361df3c489deedb51d7ca8ea64bb923b","subject":"Monkey patch original current_thread _active","status":"MERGED","created":"2020-05-07 09:42:35.000000000","updated":"2020-08-20 01:55:54.000000000","submitted":"2020-08-20 01:54:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726073-1597888465778-f63fd487","meta_rev_id":"4f94686dfb35ae51df2f05ac69ad6ed3a2a3b61f","_number":726073,"virtual_id_number":726073,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-20 01:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-08-19 22:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-08-19 22:50:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-08-19 22:50:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 09:42:35.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"REVIEWER"},{"updated":"2020-05-07 10:25:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-08 14:11:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-08-20 01:54:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5146d4746f3854d83162edcbb317fa4de78659e9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-07 09:42:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"be7df342e97c9a7f751373720296d2c7b0961c2b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-07 10:25:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"627572571de80ae9c20f056850504b4f232bc953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 11:33:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad5eb3da42174666bbf91cf8baec55cf : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d3c33273ea14b7181c29fdec1e6da32 : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b668400b3e7c438baefa2912a3ae7765 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1abcfd0d487a461e83cf3a06fc0d2778 : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8262e54813da43c3b095e36c39f9d19d : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/283683fb947d421ab244be71d6d5d584 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e0c0f61d03c481e8957cdf2bb8e0a2d : SUCCESS in 9m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/bccd133038c044e896eb030ada2afa38 : FAILURE in 1h 21m 18s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/571d0b491f86446b9338def4230741ed : FAILURE in 21m 06s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c4262d2790f74541874919aa7f2a0307 : SUCCESS in 1h 45m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c337f37efe04959ce49bb7ff63828c7e7352a27b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-07 11:41:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e226e426162b392d5b9518067d48dfb75abc0b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:58:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a63803578a584440a895f06dfbf0314b : SUCCESS in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04f5f02fb9104158a59fb4a3d94946eb : SUCCESS in 13m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df4054f9542f4afbacc2b2f54a12b854 : SUCCESS in 12m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b1555531b52451bb740453ef28f4f72 : SUCCESS in 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f630bd708f4e269e01f99fa8dd0004 : SUCCESS in 9m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/661c919cf7f64d47b7e4b530d478a185 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/940a711acdee4563bc0d7db8fdde97f4 : SUCCESS in 12m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d870d41ef2c5486fb150b4fbd358e836 : SUCCESS in 1h 20m 07s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/5bc24710a82b4420a18d0e6defe045dc : FAILURE in 16m 05s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/d32e30c809664d82acb23d3cd2cd9e69 : SUCCESS in 2h 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7d436c1be40792c395726d957267c0bdef375403","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-08 14:11:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ddf38d6002b2b5da247704b3c04f00863d0ba8e1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-19 14:00:15.000000000","message":"Patch Set 1: Workflow+1\n\nSimple fix applied in many repositories (and backports), approving","accounts_in_message":[],"_revision_number":1},{"id":"b1397e3ed7fc7c49c86eda1bd54f2a59c7e61192","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 14:00:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bab947807a88257370a990ceebd3ddd632c993d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 15:37:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61d12835f66845f080eb745bf6ad8dbd : FAILURE in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fbb53f087a34f17a7d3548f1f53e550 : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7893e284762047399c00d59bc6dee682 : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b99fa7d63e4e4cb6eb37de5a88b4ab : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80d91965ed924baa88c63c5ba2d8425e : SUCCESS in 6m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/d09694c7689841f89fb1cb26cd5e347d : SUCCESS in 1h 20m 02s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/e45b2c6eafb245cd90250e5f80a589a8 : SUCCESS in 1h 22m 21s","accounts_in_message":[],"_revision_number":1},{"id":"cc3e0f14a26d2fd375a53b828264eb9aa13c0e15","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-19 22:50:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fa854cc894d23f03c3987baa357f99dde63a726a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-19 22:50:30.000000000","message":"Patch Set 2: Workflow+1\n\nJust rebasing on pep8 fix","accounts_in_message":[],"_revision_number":2},{"id":"aa0b3c09951f713a9b727dc5d8b93b39d9dde9f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 00:19:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38152a6a6e494b3588c1b9515b740f22 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95c6b71e2cf54be4bd00b92aea76d691 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/879cab26fe6f4a9bb69063aec7a10cc2 : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/176eec700ff542e083f83c789a8c7356 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5841b9ce07c64ac384d872c8a9aab63e : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15dbc07226fb4c05af777a9d71af2fc8 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f5f79d54f684d7a8dd7840e37652099 : SUCCESS in 9m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/13f9d320ec2e446dbb4e40f05b8d5a3d : SUCCESS in 1h 19m 50s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/9316401b25e14a68b97b1c36826c0316 : FAILURE in 20m 46s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3ba98efa03e54cfc909af8a8e17698d9 : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":2},{"id":"64622102ecbcb47d79de04e7317a74668eddb462","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 00:19:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b16728d46a7566372591f927cc618007527db84f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 01:54:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8662e652f13d4e3d8511007d2e4d20f3 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d39a2d99c1d246bcbd239db5fe353093 : SUCCESS in 8m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bca7284fd19f43a08900b1ce57da9325 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44db8d3c7773460e97b835daeb271a38 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5162ff019e8543ba92a7d6429a9181e9 : SUCCESS in 8m 38s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/ce6c5333f6b9451e9544a89ca5007f1b : SUCCESS in 1h 14m 22s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c3bd164da03f4a1681ee48b630e6ddc2 : SUCCESS in 1h 26m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7056d8838199e28e58928e7d0893d74ea3b63821","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 01:54:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3b4e310d56a28f08d263e434eec331d395ee94cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 01:55:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9b2661c9334d6da85261ebe3ad73a8 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e866eed9f6c3e545a2fd6c0b680a718996661a5","revisions":{"7b98c0f2312f1afadf5fd94ae387baf530b4a402":{"kind":"REWORK","_number":1,"created":"2020-05-07 09:42:35.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/73/726073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/73/726073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/73/726073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/73/726073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/73/726073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/73/726073/1"}}},"commit":{"parents":[{"commit":"c63b726280e2eddb7429c571a433a8b25c5a3890","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c63b726280e2eddb7429c571a433a8b25c5a3890"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 18:18:49.000000000","tz":-240},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-05-07 09:42:35.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I5ea4bb27361df3c489deedb51d7ca8ea64bb923b\nCloses-Bug: #1863021\n(cherry picked from commit f0db421843f1161f4daa1e08b83eb2b11e78d9dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7b98c0f2312f1afadf5fd94ae387baf530b4a402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7b98c0f2312f1afadf5fd94ae387baf530b4a402"}]},"branch":"refs/heads/stable/ussuri"},"0e866eed9f6c3e545a2fd6c0b680a718996661a5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-19 22:50:20.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/73/726073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/73/726073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/73/726073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/73/726073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/73/726073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/73/726073/2"}}},"commit":{"parents":[{"commit":"b14e767b1db4ccd9a61d06fe74ce6cbd501fc3d6","subject":"Fix pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b14e767b1db4ccd9a61d06fe74ce6cbd501fc3d6"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 18:18:49.000000000","tz":-240},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-08-19 22:50:20.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I5ea4bb27361df3c489deedb51d7ca8ea64bb923b\nCloses-Bug: #1863021\n(cherry picked from commit f0db421843f1161f4daa1e08b83eb2b11e78d9dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0e866eed9f6c3e545a2fd6c0b680a718996661a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0e866eed9f6c3e545a2fd6c0b680a718996661a5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
