)]}'
{"id":"openstack%2Fnetworking-bagpipe~726782","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fussuri~I37e5b00f7cc3061c51ae3c21158e2d42124fb799","project":"openstack/networking-bagpipe","branch":"stable/ussuri","topic":"ussuri-release","hashtags":[],"change_id":"I37e5b00f7cc3061c51ae3c21158e2d42124fb799","subject":"Switch to stable/ussuri neutron-tempest-plugin jobs","status":"MERGED","created":"2020-05-11 11:28:41.000000000","updated":"2020-05-18 10:57:10.000000000","submitted":"2020-05-18 10:54:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726782-1589799240907-813c8a03","meta_rev_id":"0396f4236b67c55544822379ffdf6c0fcf70d4f6","_number":726782,"virtual_id_number":726782,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"date":"2020-05-15 18:06:07.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 10:54:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-05-14 16:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-05-12 11:47:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-05-15 15:38:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-05-15 15:38:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-05-13 15:18:34.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 11:47:48.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-05-13 15:18:34.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-05-14 16:39:38.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-05-15 15:38:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-05-18 10:54:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f79c6e9180f7ef5f828ab04711d99cf42bbfe1a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-11 11:28:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16e75143b12ccf15b49bfc6770fdd7d8f00f9da9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:21:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489fceaf204c4774a20e62a9d85ab54b : SUCCESS in 10m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a44693269e904a1bbd9f98d334045d09 : SUCCESS in 10m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80923ce9224c4a598f95e9307890d858 : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7eb7bd9078f745f88689cc027a32a67c : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861e434c0ed84835b037fac7687189b1 : SUCCESS in 9m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3094279b95b24ca7a8094e52fae983ba : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91bb053deba44bfbaa9744323351be6d : SUCCESS in 9m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/5bd96de8ab9147a282a1fe022464ac90 : SUCCESS in 1h 32m 37s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/8be609d9aa104f56b081e93505948f43 : FAILURE in 22m 48s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/29246ea94cbb4c598d71fa72e913ae00 : SUCCESS in 1h 42m 33s","accounts_in_message":[],"_revision_number":1},{"id":"310161ef8ff13c0c3bb1c985e285210541c538a7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-12 11:47:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d2da36a9a8db9e0b02f1e5255385b8aa75b4e34","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-14 16:39:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"450d26aabe77f513db44d89f66a0fc98cc014fce","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-15 11:55:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73bf80541f1a278a358dfc561a240351fb8d5f38","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-15 15:38:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"762b3dcc339affef2af3772dcb3b4f9b61ee18c3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-15 18:06:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e5660d412fd68340e818ba30974732e0eb188bc6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-18 07:31:57.000000000","message":"Patch Set 1:\n\nrecheck Depends-on is merged","accounts_in_message":[],"_revision_number":1},{"id":"a93794101406c73cc4d21a81a039e842bae5f261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 09:12:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7928dd7f1af44742b3a58f0c3bb45c92 : SUCCESS in 9m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e410a69996c84d6f9c2ba6697467e064 : SUCCESS in 9m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/363d95226c5742dfa295533c72b173cc : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bdd917018e842449540a5c93e424768 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd245df1922c469894b34e320da89418 : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4f1c4ad00be4a77a5fb51cab2d9316a : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f5087d9d7a542da9de5a8d7446e431a : SUCCESS in 8m 26s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/71741fcdee2b418780942b194816c97b : SUCCESS in 1h 32m 29s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/fab3000cb4714a0390b618a83cc9b203 : FAILURE in 19m 26s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/713c99414f4c43fd970a1888ce1325b8 : SUCCESS in 1h 35m 04s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":1},{"id":"ae4e3841dd06906b0b6dff6f1804caa075e23e98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 09:12:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93968a6e81b10b59f3602e7289437e740c876cac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 10:54:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b13158827df4032bfe077aaac8529c0 : SUCCESS in 10m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1e751b0db14407eb327c6f34b648378 : SUCCESS in 8m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/787f0ab576fb4b868aed184cc8c43cb2 : SUCCESS in 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0b2345bba94cee8768a207ea9ad17c : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b92bcf033c4544b9b9b2e19b81357fa3 : SUCCESS in 9m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/53c3cd75b60646259e9e02f0e6560afb : SUCCESS in 1h 35m 40s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3acfc3d70f9b44d897bd2b11196beb80 : SUCCESS in 1h 22m 23s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":1},{"id":"f3dc212058546f9c919da30340f9ecb8fcb7c243","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 10:54:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"80eec8ba0d542f968c3f9f0f4080873c355d57d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 10:57:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f0846df6194f51aa61e71c6aca76ea : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6bee053b2523f94ec621f83237f02acafb42e27","revisions":{"d6bee053b2523f94ec621f83237f02acafb42e27":{"kind":"REWORK","_number":1,"created":"2020-05-11 11:28:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/82/726782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/82/726782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/82/726782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/82/726782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/82/726782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/82/726782/1"}}},"commit":{"parents":[{"commit":"c63b726280e2eddb7429c571a433a8b25c5a3890","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c63b726280e2eddb7429c571a433a8b25c5a3890"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-05-11 11:28:22.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-05-11 11:28:22.000000000","tz":120},"subject":"Switch to stable/ussuri neutron-tempest-plugin jobs","message":"Switch to stable/ussuri neutron-tempest-plugin jobs\n\nDepends-On: https://review.opendev.org/726776\nChange-Id: I37e5b00f7cc3061c51ae3c21158e2d42124fb799\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d6bee053b2523f94ec621f83237f02acafb42e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d6bee053b2523f94ec621f83237f02acafb42e27"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
