)]}'
{"id":"openstack%2Fnetworking-bagpipe~729872","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fstein~Idccc749f154ef1cb71c0f02bf41825a197faa0b2","project":"openstack/networking-bagpipe","branch":"stable/stein","topic":"fix_periodic_docs","hashtags":[],"change_id":"Idccc749f154ef1cb71c0f02bf41825a197faa0b2","subject":"[Stable only]: make stable tox-docs job passing on stein","status":"ABANDONED","created":"2020-05-21 08:40:45.000000000","updated":"2020-05-21 12:08:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0f9505dbdb528ec8026df8944ff70dbfe5929f6","_number":729872,"virtual_id_number":729872,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-21 12:07:59.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-21 09:45:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 09:45:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1844abe23e2216c8ca47b6290d859fb8ecfa50b2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-21 08:40:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84a6bd7164e4ed9c9545913598e83212d8d802bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 09:45:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f870eff798f48c3917ff8be60b8edda : SUCCESS in 8m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fedaee16a3bb481193b262286beac1e8 : SUCCESS in 10m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/20b33359b5b349d1ba5ebf1e111dfcec : SUCCESS in 11m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdd4088dc6314bd8a6b1bcf97e63f36a : SUCCESS in 8m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc9ab28ddbc24025aa79a8f0ee9b65e7 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a71e47015946fc8d06112f2ec12ada : FAILURE in 7m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca63fa50c3204ab3bbcbdf6bbdd16669 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e1dfd3db3624fae853ccb466a1ae163 : SUCCESS in 12m 01s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/aac826ac5cec45538252068e2d3e0f58 : SUCCESS in 1h 02m 42s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7f4ece3d27464eafb91314df097ca06b : FAILURE in 34m 01s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/908132cb7d41458ca909fb74f1454ecc : FAILURE in 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba755c32e3f8cdc638859670c4aa55cc8d3fd961","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-21 12:07:59.000000000","message":"Patch Set 1:\n\nThis one: https://review.opendev.org/729936 seems more promising","accounts_in_message":[],"_revision_number":1},{"id":"00e0f0a1925a5b1b251b39f8c62982943b1f8199","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-21 12:08:28.000000000","message":"Abandoned\n\nThis one seems to be the good solution: https://review.opendev.org/729936","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0e90425f45d5c72aa28884e8e595697e22d315c","revisions":{"e0e90425f45d5c72aa28884e8e595697e22d315c":{"kind":"REWORK","_number":1,"created":"2020-05-21 08:40:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/729872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/729872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/729872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/729872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/729872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/729872/1"}}},"commit":{"parents":[{"commit":"e8d2fca2a498e5efc5a925377dc17dca3efeeb4e","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e8d2fca2a498e5efc5a925377dc17dca3efeeb4e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-05-21 08:36:18.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-05-21 08:36:18.000000000","tz":120},"subject":"[Stable only]: make stable tox-docs job passing on stein","message":"[Stable only]: make stable tox-docs job passing on stein\n\nstable/stein docs building started to fail recently with error like\nthis:\nException occurred:\n.../networking_bgpvpn/neutron/db/bgpvpn_db.py\", line 30, in \u003cmodule\u003e\n  from neutron_lib.db import standard_attr\nImportError: cannot import name \u0027standard_attr\u0027\n\nFix networking-bgpvpn in required-projects list for docs job definition.\n\nChange-Id: Idccc749f154ef1cb71c0f02bf41825a197faa0b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e0e90425f45d5c72aa28884e8e595697e22d315c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e0e90425f45d5c72aa28884e8e595697e22d315c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
