)]}'
{"id":"openstack%2Fnetworking-bagpipe~729936","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fstein~I599ba9823bbf3abc770cb9893489f42c7dbf477e","project":"openstack/networking-bagpipe","branch":"stable/stein","topic":"bagpipe-stein-gate-fix","hashtags":[],"change_id":"I599ba9823bbf3abc770cb9893489f42c7dbf477e","subject":"[stein-only] Fix networking-bgpvpn and neutron requirement constraints","status":"MERGED","created":"2020-05-21 12:04:16.000000000","updated":"2020-05-26 17:50:30.000000000","submitted":"2020-05-26 17:47:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729936-1590515232172-231207c7","meta_rev_id":"5a806e18c2fddb5772e58b9bc75f919e5addb9cc","_number":729936,"virtual_id_number":729936,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-26 15:00:24.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 17:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-05-26 08:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-05-25 07:42:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-05-25 07:38:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-05-26 08:06:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 17:41:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-25 07:38:50.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-05-25 07:42:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-26 08:06:38.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-05-26 17:47:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2aa58a74ea785ca4a89048a6f1c14b6cd40351a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-21 12:04:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac547bb1deafe1593f12cca1a992606cc4494751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 12:26:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/423116a0c9bc47fdad3c19b42ced60dd : SUCCESS in 8m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/508799d8b0574cb3909473161b9d189e : SUCCESS in 7m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6a3754e46eeb45ffb98eafa72290b707 : SUCCESS in 9m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07a9961d5b64487ab05b85022a23f3c7 : SUCCESS in 10m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6251cbcf82464969883bb3dcf7cfcef5 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec02dd9a996a47bbabc3313fdd2383a7 : SUCCESS in 6m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/883bf5cd08124f339bc42607d9ea24c3 : FAILURE in 4m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aed71eb4c28641ea8590428aa542fa9f : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/989be7aa977f4fefaec43dc410f75e9d : SUCCESS in 9m 55s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/8471e90177a441059049ea596cd87fe2 : FAILURE in 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef110a7fe59d9590407056406b958526d3589971","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-21 14:03:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c520a185268038ae21381507046c9b3a4fc3a9c6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-21 14:12:16.000000000","message":"Topic set to bagpipe-stein-gate-fix","accounts_in_message":[],"_revision_number":2},{"id":"2405da983e0d598b1783859cf27902242e3bc28e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 15:02:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31dca65acff94815bf85b2d7f3672775 : SUCCESS in 8m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82d52799bf324122ae4bc8860cab99c0 : SUCCESS in 8m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/567ff52def674ca4b02fb453feb46816 : SUCCESS in 9m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f818769e8f6940c0811963156c6d72b6 : SUCCESS in 11m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f88176b232094ab2aa39009d6c5a1f2d : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0be6081c8dd44e9ad6bde8673adc88f : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f5fc55b661b4d79b18fa766851e052d : SUCCESS in 5m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a43df0b442b47938979f7ea2dede5d2 : SUCCESS in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56ec8de4f1e3417cbf02dfed7425c24d : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4be9ccc3a53a46b1bda14605949f7efe : SUCCESS in 9m 29s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/c8271f51063f4146b711edc3f8ffe4f2 : SUCCESS in 51m 58s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/edbad6500599414aac02f2d539be5781 : FAILURE in 39m 53s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/cf0936fea6014aedb11898412f6e0c60 : FAILURE in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df6d6cb9bf512e3af4fafd4556ffb4834cf7ca1d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-21 15:06:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"77befb3cc1a80d9b13c0a2018a666bdb1c1901d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 16:02:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/416fd83642134f38b1602fe508114c0a : SUCCESS in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f7fb43878824eab9a6a2e4d615fb34d : SUCCESS in 8m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2e61fcff44bb4a76af9169730374095a : SUCCESS in 11m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0b4d0f99b2641108121d44e82e7f836 : SUCCESS in 8m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f53dd2c08b04067977d591427624bf6 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcba9ace610140f188c7d8caa75f30a7 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2da7cfc4aca74abdb0fa607c2dcb30e9 : SUCCESS in 5m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/21dabf6a315a4623ad390ca8d05b8c90 : SUCCESS in 2m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a049fa10c584db49beac370d0cbd114 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c339b4eafb06440c94ff94ae484e79fb : SUCCESS in 8m 53s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/c3c682e102414889b7d6d73dba1600d1 : SUCCESS in 52m 56s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7d3f36f464bb4642bb1647f4b0b8e91f : FAILURE in 28m 47s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/b7a2bf7247a84bd1b71aaa59a8907de9 : FAILURE in 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f89f4f21077361e7acfe96e44749b28f6169e8bc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-22 06:14:14.000000000","message":"Patch Set 2:\n\nThanks for working on this\nAs legacy-tempest-dsvm-networking-bgpvpn-bagpipe fails with:\nThe specified regex doesn\u0027t match with anythingERROR: InvocationError for command /opt/stack/new/tempest/.tox/all-plugin/bin/tempest run --regex \u0027^networking_bgpvpn_tempest\\.\u0027 --concurrency\u003d4 (exited with code 1)\n\nMy idea would be to use regex from neutron-tempest-plugin, like:\nneutron_tempest_plugin.bgpvpn","accounts_in_message":[],"_revision_number":2},{"id":"afbaad36a6a1339b8d643206d8851e8c4d49c1f3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-22 06:22:51.000000000","message":"Patch Set 2:\n\n\u003e Thanks for working on this\n \u003e As legacy-tempest-dsvm-networking-bgpvpn-bagpipe fails with:\n \u003e The specified regex doesn\u0027t match with anythingERROR:\n \u003e InvocationError for command /opt/stack/new/tempest/.tox/all-plugin/bin/tempest\n \u003e run --regex \u0027^networking_bgpvpn_tempest\\.\u0027 --concurrency\u003d4 (exited\n \u003e with code 1)\n \u003e \n \u003e My idea would be to use regex from neutron-tempest-plugin, like:\n \u003e neutron_tempest_plugin.bgpvpn\n\nNo, it\u0027s defined here:\nhttps://opendev.org/openstack/networking-bgpvpn/src/branch/stable/stein/devstack/devstack-gate-bagpipe-rc#L7","accounts_in_message":[],"_revision_number":2},{"id":"28ad08468ee3cfc3096d7033bd276b63bc0f51e4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 17:05:43.000000000","message":"Patch Set 2:\n\nlet\u0027s try with this fix -https://review.opendev.org/#/c/730378/1","accounts_in_message":[],"_revision_number":2},{"id":"e3b66e99119d99fa11acfb2d8f02ee869a320180","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-22 17:35:04.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4d80236e1431ee8e3e44ec7b77177b70b758a113","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-22 17:35:53.000000000","message":"Patch Set 3:\n\n\u003e let\u0027s try with this fix -https://review.opendev.org/#/c/730378/1\n\nThanks! Depends-On tag added!","accounts_in_message":[],"_revision_number":3},{"id":"acc26faab0e92b01541cbb13fbcce7fff68cc6c5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 17:41:19.000000000","message":"Patch Set 3:\n\n\u003e \u003e let\u0027s try with this fix -https://review.opendev.org/#/c/730378/1\n \u003e \n \u003e Thanks! Depends-On tag added!\n\nneed one more fix on bgpvpn side. can you make depends-on with https://review.opendev.org/#/c/730379/ which internal depends-on on zuul job fix","accounts_in_message":[],"_revision_number":3},{"id":"89e94e11a4e82a8623399bec4d61262cfecd8986","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-22 17:43:36.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"838fe9d996f5318cbdb17501b13fa36e088da1b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 19:06:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969efd3cac444f648a5b659177c8c88e : SUCCESS in 7m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c311b460fb284dc0ad6c621a20574752 : SUCCESS in 10m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/79526f657b3d426aba5b9b684c06f92d : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46f97beae4a04432a8cfe14f8d4e4aff : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db9662501bc24806bb4c51e2c1fd297b : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/655378c0f0a54e34ab37bea8c7d1c282 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/589ebbbbbc9b4e9ab932625166d27cc1 : SUCCESS in 4m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/36dadbb30bc643dba81220e24ed50a52 : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d217199e676e490ebbacfccca252a49c : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c28c14bcc9304cbeb4691b16a4eff04b : SUCCESS in 8m 20s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/eea2e3be377d4b45bd64aaca060b9b5a : SUCCESS in 53m 49s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9db603a68ac4478793f4f9e1de441aa6 : SUCCESS in 1h 20m 30s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/bd13c59a8b644fa3a9d74282357fd44e : FAILURE in 13m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"504cbe026e243a081ba5b85e9aff41aabe496d23","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-22 19:35:25.000000000","message":"Patch Set 4:\n\n\u003e \u003e \u003e let\u0027s try with this fix -https://review.opendev.org/#/c/730378/1\n \u003e \u003e\n \u003e \u003e Thanks! Depends-On tag added!\n \u003e \n \u003e need one more fix on bgpvpn side. can you make depends-on with\n \u003e https://review.opendev.org/#/c/730379/ which internal depends-on on\n \u003e zuul job fix\n\nIt\u0027s working!\n\nhttps://zuul.opendev.org/t/openstack/build/9db603a68ac4478793f4f9e1de441aa6/log/job-output.txt#2411","accounts_in_message":[],"_revision_number":4},{"id":"a4fafdc2ecbf084bec5f4e07031e5c00ef6c2884","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-25 07:38:50.000000000","message":"Patch Set 4: Code-Review+1\n\nThank you Elod and Gmann for working on this","accounts_in_message":[],"_revision_number":4},{"id":"fc20205ec337b2f251e93f9222fe535ac44ce8ef","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-25 07:42:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"34385efce814b4b4d139b9fbb53a48acefa31c18","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-26 08:06:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLooks good indeed thanks!","accounts_in_message":[],"_revision_number":4},{"id":"8c92b2c3548e0f4dc179e3a26f6c783f599bfed6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-26 15:00:24.000000000","message":"Patch Set 4:\n\nrecheck - patch that this is depends on merged. gate job should have been started.","accounts_in_message":[],"_revision_number":4},{"id":"7fa4d92202fe909a9617ee055bdce072a1e12436","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:14:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52d41c2f5eb845e29b1dcb09eb86f301 : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9705daf218c34ac798a25bf3f5e9c712 : SUCCESS in 8m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b8f40f25deb84d08b63e545ca6f6162f : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/857e0b9ffac74fc2958066ba7853e4bd : SUCCESS in 11m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55fb2f7d9fd64dfa9f15c43ead638877 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf48238e2a54a59935eab1dc9012c23 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4f14d24e78f425298f220a7843ca50b : SUCCESS in 7m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e86b68fa38f4b96884d5ddbc7f4231b : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34cfcb54c4e241c6b6be9b128b43737f : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d02630ef6f542ebbc85405520c29777 : SUCCESS in 9m 48s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/a6196a4972e74b24baf439eb657a76df : SUCCESS in 1h 06m 29s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5cc037c4ee7d46508fa993820e4c4912 : SUCCESS in 1h 09m 04s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/6a6884e4d0404b138e9ebf69e8a386f3 : FAILURE in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"af23bad71d66d3808cc0a6658f08720f207ac29e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:14:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f24e096da40fa20c74ff3b2142a81b47cc10e0c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:47:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffa6bd55d09d43d286ac08b6f6d9b696 : SUCCESS in 8m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ca665c0728f9462fb45e5152a79bb646 : SUCCESS in 8m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bce28ad0fb16492390272cf0cf9df026 : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/501f041ae5b6465782efe7aaaf59821d : SUCCESS in 9m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/955a0042d6984f6cba8c2d526cb7f643 : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc007f9a8ab14e5eb5b829f367a4c5e5 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e91c21d9fed74c8fb517da995cce7276 : SUCCESS in 5m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/600bd328860c4e9a88798df5d5f584ce : SUCCESS in 4m 52s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/1ceeb2fbdc314b26aacecc3b440a9f92 : SUCCESS in 57m 34s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e361f78f1f104a92bf4f4c16e2309142 : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":4},{"id":"92a3336eeca7a47d34cc22579e91b1b3c7ed31b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:47:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5a806e18c2fddb5772e58b9bc75f919e5addb9cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:50:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007add9aa3af41f38d097dd4530690c3 : SUCCESS in 2m 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faf4b1ef423a48a49256d816dc450258 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a1b587befdac648e168e02d052af6c412f53b523","revisions":{"bbee390a08ce8798f2681667168ca2b6a0aa4feb":{"kind":"REWORK","_number":1,"created":"2020-05-21 12:04:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/36/729936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/36/729936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/1"}}},"commit":{"parents":[{"commit":"e8d2fca2a498e5efc5a925377dc17dca3efeeb4e","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e8d2fca2a498e5efc5a925377dc17dca3efeeb4e"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-05-21 11:44:44.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-05-21 11:47:50.000000000","tz":120},"subject":"[stein-only] Fix networking-bgpvpn requirement constraint","message":"[stein-only] Fix networking-bgpvpn requirement constraint\n\nUssuri version of networking-bgpvpn (12.0.0) was released and with that\nversion of networking-bgpvpn tox -e docs started to fail when tox\ninstalls latest bgpvpn. Let\u0027s constrain networking-bgpvpn to Stein\nversions.\n\nChange-Id: I599ba9823bbf3abc770cb9893489f42c7dbf477e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/bbee390a08ce8798f2681667168ca2b6a0aa4feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/bbee390a08ce8798f2681667168ca2b6a0aa4feb"}]},"branch":"refs/heads/stable/stein"},"bab6c76a9f0454db9255982a3259f7fc4183e208":{"kind":"REWORK","_number":2,"created":"2020-05-21 14:03:13.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/36/729936/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/36/729936/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/2"}}},"commit":{"parents":[{"commit":"e8d2fca2a498e5efc5a925377dc17dca3efeeb4e","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e8d2fca2a498e5efc5a925377dc17dca3efeeb4e"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-05-21 11:44:44.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-05-21 13:52:17.000000000","tz":120},"subject":"[stein-only] Fix networking-bgpvpn and neutron requirement constraints","message":"[stein-only] Fix networking-bgpvpn and neutron requirement constraints\n\nUssuri version of networking-bgpvpn (12.0.0) was released and with that\nversion of networking-bgpvpn tox -e docs started to fail when tox\ninstalls latest bgpvpn. Let\u0027s constrain networking-bgpvpn to Stein\nversions. This needs to be done in requirements.txt, since\nnetworking-bgpvpn is not part of upper-constraints.txt.\n\nNow that requirements.txt is modified the check job finds problem with\nneutron constraints as well. The problems comes due to neutron was\nmeanwhile added to upper-constraints.txt. This patch fixes neutron\u0027s\nrequirements to make the check job pass.\n\nDocs tox target\u0027s dependencies had to be updated to make use of the\nupper-constraints.txt. However sphinx installs bagpipe and its\nrequirements.txt it does not use the upper constraints so let\u0027s\ninstall the requirements earlier with the \u0027deps\u0027.\n\nChange-Id: I599ba9823bbf3abc770cb9893489f42c7dbf477e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/bab6c76a9f0454db9255982a3259f7fc4183e208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/bab6c76a9f0454db9255982a3259f7fc4183e208"}]},"branch":"refs/heads/stable/stein"},"7dc1771388144ac3cb9de4304b542fa40f87e6d4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-22 17:35:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/36/729936/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/36/729936/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/3"}}},"commit":{"parents":[{"commit":"e8d2fca2a498e5efc5a925377dc17dca3efeeb4e","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e8d2fca2a498e5efc5a925377dc17dca3efeeb4e"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-05-21 11:44:44.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-05-22 17:34:56.000000000","tz":0},"subject":"[stein-only] Fix networking-bgpvpn and neutron requirement constraints","message":"[stein-only] Fix networking-bgpvpn and neutron requirement constraints\n\nUssuri version of networking-bgpvpn (12.0.0) was released and with that\nversion of networking-bgpvpn tox -e docs started to fail when tox\ninstalls latest bgpvpn. Let\u0027s constrain networking-bgpvpn to Stein\nversions. This needs to be done in requirements.txt, since\nnetworking-bgpvpn is not part of upper-constraints.txt.\n\nNow that requirements.txt is modified the check job finds problem with\nneutron constraints as well. The problems comes due to neutron was\nmeanwhile added to upper-constraints.txt. This patch fixes neutron\u0027s\nrequirements to make the check job pass.\n\nDocs tox target\u0027s dependencies had to be updated to make use of the\nupper-constraints.txt. However sphinx installs bagpipe and its\nrequirements.txt it does not use the upper constraints so let\u0027s\ninstall the requirements earlier with the \u0027deps\u0027.\n\nDepends-On: https://review.opendev.org/#/c/730378\nChange-Id: I599ba9823bbf3abc770cb9893489f42c7dbf477e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7dc1771388144ac3cb9de4304b542fa40f87e6d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7dc1771388144ac3cb9de4304b542fa40f87e6d4"}]},"branch":"refs/heads/stable/stein"},"a1b587befdac648e168e02d052af6c412f53b523":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-22 17:43:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/36/729936/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/36/729936/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/36/729936/4"}}},"commit":{"parents":[{"commit":"e8d2fca2a498e5efc5a925377dc17dca3efeeb4e","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e8d2fca2a498e5efc5a925377dc17dca3efeeb4e"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-05-21 11:44:44.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-05-22 17:43:33.000000000","tz":0},"subject":"[stein-only] Fix networking-bgpvpn and neutron requirement constraints","message":"[stein-only] Fix networking-bgpvpn and neutron requirement constraints\n\nUssuri version of networking-bgpvpn (12.0.0) was released and with that\nversion of networking-bgpvpn tox -e docs started to fail when tox\ninstalls latest bgpvpn. Let\u0027s constrain networking-bgpvpn to Stein\nversions. This needs to be done in requirements.txt, since\nnetworking-bgpvpn is not part of upper-constraints.txt.\n\nNow that requirements.txt is modified the check job finds problem with\nneutron constraints as well. The problems comes due to neutron was\nmeanwhile added to upper-constraints.txt. This patch fixes neutron\u0027s\nrequirements to make the check job pass.\n\nDocs tox target\u0027s dependencies had to be updated to make use of the\nupper-constraints.txt. However sphinx installs bagpipe and its\nrequirements.txt it does not use the upper constraints so let\u0027s\ninstall the requirements earlier with the \u0027deps\u0027.\n\nDepends-On: https://review.opendev.org/#/c/730379\nChange-Id: I599ba9823bbf3abc770cb9893489f42c7dbf477e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a1b587befdac648e168e02d052af6c412f53b523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a1b587befdac648e168e02d052af6c412f53b523"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
