)]}'
{"id":"openstack%2Fnetworking-bagpipe~729938","triplet_id":"openstack%2Fnetworking-bagpipe~master~Idea5a0594fbad2abdcbf7116dcad4e4ea0fca4d9","project":"openstack/networking-bagpipe","branch":"master","hashtags":[],"change_id":"Idea5a0594fbad2abdcbf7116dcad4e4ea0fca4d9","subject":"Add requirements.txt to docs deps","status":"MERGED","created":"2020-05-21 12:16:46.000000000","updated":"2020-10-08 21:34:44.000000000","submitted":"2020-10-08 21:30:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729938-1602192646958-3e89e6fb","meta_rev_id":"3215e591a64df6bc93e265ae02498cfeea447604","_number":729938,"virtual_id_number":729938,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-25 10:16:09.000000000","_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"date":"2020-05-25 10:39:05.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-08 21:30:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-10-08 20:01:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-10-08 20:01:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-05-22 12:01:37.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 12:01:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-05-25 10:16:09.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-10-08 20:01:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-08 21:30:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"780443332bd92ecd2ec6b245ad434da64179acdd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-21 12:16:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e341bb7e50bd89468d6df76f56ab81d278b8673c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 13:51:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3da981f4d4544de7a8c6183d91af1de7 : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69929407d5ef46bdb5e6d9c8cdbbfe71 : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9fdfc6d2c3a4d46bf2d6a99a15283e8 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cce878eec9d4d3eaaa941e129959022 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30bc971d32b9454fa3dc032757d4cf25 : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d9281dd95744f8096848c330dd0e292 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5174d22c8c04038b671e520447de4a2 : SUCCESS in 8m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93e097b97e914118bebbe74ee7f5f62b : SUCCESS in 9m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c686e366aab640448bbae477ba9e49bb : SUCCESS in 1h 32m 07s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f7a102c3ea404017abc2caa49cf63f83 : FAILURE in 17m 02s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ccb7630a2ad646eea23985650147cd36 : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"92701bd5699c771ceb2a6f4f4fc059c9c1e6cd3f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-22 12:01:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcd2158c04cb48ba70171e0c90b08a83e0136654","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-22 12:12:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ac32117bdb1f3119d3ef1125f2fb178a581e488","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-05-25 07:38:09.000000000","message":"Patch Set 1:\n\nA few things are surprising to me:\n* it would seem incorrect to me to add runtime requirements for \n  sphinx doc generation, which does not need all these requirements\n* I don\u0027t get how using /requirements.txt would result in using upper-constraints\n* in fact, upper-constraints.txt should be active already (line 79 of tox.ini)","accounts_in_message":[],"_revision_number":1},{"id":"2804f1351263851d1570a6feb39a5b7fc9903cfe","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-25 08:35:23.000000000","message":"Patch Set 1:\n\n\u003e A few things are surprising to me:\n \u003e * it would seem incorrect to me to add runtime requirements for\n \u003e sphinx doc generation, which does not need all these requirements\n \u003e * I don\u0027t get how using /requirements.txt would result in using\n \u003e upper-constraints\n \u003e * in fact, upper-constraints.txt should be active already (line 79\n \u003e of tox.ini)\n\nHi Thomas,\n\nTox works like this:\n\n1. it installs requirements, listed in deps, with pip (see log [1])\n2. then it runs the command, which is now sphinx. Sphinx installs bagpipe with its dependencies, listed in requirements.txt (see log [2])\n\nSo these are two separate steps, and the second could fail, because sphinx does not use any constraints (see example in Stein, where the latest bgpvpn is being installed which has conflicts with the stein version of bagpipe [3])\n\nHowever, if we add requirements.txt to the first step, then sphinx won\u0027t reinstall them and there won\u0027t be any conflict either as the correct versions are installed already.\n\nThis is actually not really a problem in master, but does not harm to have it here and in the future where we cut new stable branches there won\u0027t be a problem either in this way.\n\nI hope my explanation is somewhat clear, let me know if you have further questions.\n\n[1] https://zuul.opendev.org/t/openstack/build/5cce878eec9d4d3eaaa941e129959022/log/tox/docs-1.log\n[2] https://zuul.opendev.org/t/openstack/build/5cce878eec9d4d3eaaa941e129959022/log/sphinx-build-pdf.log#25\n[3] https://zuul.opendev.org/t/openstack/build/583e1a1e7190492aa1e9cc0ab3dc56fc/log/job-output.txt#1485","accounts_in_message":[],"_revision_number":1},{"id":"a49b673d9b95b84e8d8e8ccb5118e280e86e4a52","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-05-25 10:16:09.000000000","message":"Patch Set 1:\n\n\u003e I hope my explanation is somewhat clear, let me know if you have\n \u003e further questions.\n\nWell, it does help a bit, yes. Thanks.\n\nI hadn\u0027t in mind that doc building required the whole package installation -- that must be required for things like autogenerated config documentation to be possible.\n\nOne thing I\u0027m still unclear about is why the \"python setup.py\" step done by sphinx ([1]) does not honnor the upper version constraints set for networking-bgpvpn in requirements.txt.  Any idea if we could fix that ?\n\n\n[1] https://zuul.opendev.org/t/openstack/build/5cce878eec9d4d3eaaa941e129959022/log/sphinx-build-pdf.log#11","accounts_in_message":[],"_revision_number":1},{"id":"11740e2f89120588ac6bbae486113e33b71b05e5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-25 10:39:05.000000000","message":"Patch Set 1:\n\n\u003e One thing I\u0027m still unclear about is why the \"python setup.py\" step\n \u003e done by sphinx ([1]) does not honnor the upper version constraints\n \u003e set for networking-bgpvpn in requirements.txt.  Any idea if we\n \u003e could fix that ?\n \u003e \n \u003e \n \u003e [1] https://zuul.opendev.org/t/openstack/build/5cce878eec9d4d3eaaa941e129959022/log/sphinx-build-pdf.log#11\n\nIf there are upper constraint in requirements.txt then sphinx would use that, but where possible we should avoid using upper constraints in requirements.txt and rather use global upper-constraints.txt.\n\nActually neutron was added to global upper-constraints.txt recently, so for neutron this patch would eliminate constraint issues in the future (as upper-constraints.txt will be used).\nHowever, bgpvpn is not added to global upper-constraints.txt, so for that we have to update requirements.txt on every newly created stable branch in the future. This patch does not help on that, unfrotunately. (Sorry for using bgpvpn as example in my previous comment, as it is a wrong example)","accounts_in_message":[],"_revision_number":1},{"id":"0c5e893be3b05d1795af05a76c445224d439fe76","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-08 20:01:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"808f4e2ff9bf79a299ca932557fd09d485ef660b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 20:02:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9cf748367758a2c2f3121a3f03c3b28ed09e00a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 21:30:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"50c7c6ec35382f67929ecb67fee9837b4d865397","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 21:30:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/502e278901bb43d9a6b56e53278a14ae : SUCCESS in 9m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad4d54a80ff545c4b674698946d2c1f5 : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d27d039afdf46459f3616054b24638a : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fac90a021ba4c2a87223f23491ed35f : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fac242a7ec954b71ade3ab5dc5896a1a : SUCCESS in 4m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/26ab37b714f34dd3b11bc93e3e709c3d : SUCCESS in 1h 20m 46s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b218ccc9f7f5480b8a5d7b4f7efbe6f1 : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3215e591a64df6bc93e265ae02498cfeea447604","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 21:34:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bf700841af4fc48711b9bf30cafd58 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f560657090c4a43afc3fbef56f4762e : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fa00945f42a2cd19d5e1d5a229fed28853276ae","revisions":{"4fa00945f42a2cd19d5e1d5a229fed28853276ae":{"kind":"REWORK","_number":1,"created":"2020-05-21 12:16:46.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/38/729938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/38/729938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/38/729938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/38/729938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/38/729938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/38/729938/1"}}},"commit":{"parents":[{"commit":"53c5909c6c42b46efa07205a7938f61b0c5489af","subject":"Merge \"Remove the dependency on the \"mock\" package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/53c5909c6c42b46efa07205a7938f61b0c5489af"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-05-21 12:12:44.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-05-21 12:12:44.000000000","tz":120},"subject":"Add requirements.txt to docs deps","message":"Add requirements.txt to docs deps\n\nSphinx installs bagpipe with all its requirements, but it does not use\nthe upper constraints. If we add the requirements.txt to docs target\u0027s\ndependencies then it will be installed with the upper constraints.\n\nChange-Id: Idea5a0594fbad2abdcbf7116dcad4e4ea0fca4d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4fa00945f42a2cd19d5e1d5a229fed28853276ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4fa00945f42a2cd19d5e1d5a229fed28853276ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
