)]}'
{"id":"openstack%2Fnetworking-bagpipe~730853","triplet_id":"openstack%2Fnetworking-bagpipe~master~If01ae1c8b16a4301506fa6cd1e893c4176a0aaa0","project":"openstack/networking-bagpipe","branch":"master","topic":"remove-six","hashtags":[],"change_id":"If01ae1c8b16a4301506fa6cd1e893c4176a0aaa0","subject":"Remove usage of six library","status":"MERGED","created":"2020-05-26 15:24:36.000000000","updated":"2020-06-03 07:56:10.000000000","submitted":"2020-06-03 07:54:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730853-1591170857045-2329ea62","meta_rev_id":"18d3b6613f514d68c54968a2893f275df3111077","_number":730853,"virtual_id_number":730853,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"date":"2020-06-02 11:42:29.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 07:54:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2020-06-03 06:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-06-02 15:31:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":1,"date":"2020-06-03 06:32:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 15:31:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-03 06:32:11.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-06-03 07:54:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6cff64912c3798ec8f71f5a176b24727ce78b8a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-26 15:24:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c642b3c8643f724c84365c82fd1e2fdf16498393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:58:10.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71705541cefe47879777b940ac424075 : FAILURE in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3641899b928a4507a9dcba8f6fca1332 : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b3dc06371204962ad32926ad9ff3317 : SUCCESS in 12m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe480b8f4f048188c899185442aa8a8 : SUCCESS in 7m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ee09a9a3d5a94002abc1d213ac6fffd2 : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5411c3e63bf42bc98daab4270525540 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47f5792648aa486ab680f3167ca19ae9 : SUCCESS in 9m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6c2ae3e5bb441a2bfc22b40e9fbf41c : SUCCESS in 10m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3d203943e6a74d3ea841d57db99954b3 : SUCCESS in 1h 21m 55s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/97af706e94384430a6edfdd96cbd9ca1 : FAILURE in 21m 44s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/5ca666c2dc83434eb56778ed9cbf08f6 : SUCCESS in 1h 16m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b1856ccb82781b10df9e2d6e68e3975b7725be38","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-27 11:02:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7a464a4a2db11aa99c0b5ae6650772c23d3f2d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 12:49:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953e3c2d6a1641b5b09a9b869d0889a2 : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf3fbff9465b48668e4513e151e9d55a : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53ec43a114984f67ab42e570cca0825d : SUCCESS in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c4a68239fc34470a4e1fc01eda79a71 : SUCCESS in 9m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/954ca286a21645eb8e73fafd62f71b8a : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/044caf54b06e41ba9f6b03233da360e7 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6447a9b5ae1543bfbf6b56157f95392c : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae6bcfbcc1d94b969c7bec13437ac87d : SUCCESS in 10m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ad06b807262b49dfbc5900baef745dc0 : SUCCESS in 1h 18m 50s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/15ae9d383ae046208fbcb5f5719c9f91 : FAILURE in 20m 20s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/255a61db381a4649a715d2f596559ad8 : SUCCESS in 1h 43m 58s","accounts_in_message":[],"_revision_number":2},{"id":"64c73af9c9bb6fd8880556cffe57f2c4694bf86d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-29 20:17:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ff1ec0a3059b523479905719d553939482522c38","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-01 06:01:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"968e2305951f385897d99e803ad249503d289743","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-02 11:39:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b19e0559c50347ab38f9389386f2bd74328cbf2f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-02 11:42:29.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"34e0dd5356a179055364891bb4070eb6817dfe2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 13:15:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be91426701c45dfb7e3c4e93a707158 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caae9ca43618402ba2b1c31d9df87692 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dab43c5aac7642c9a6e228079f02d807 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aca3e7f432241a2a8140fb1010e1b44 : SUCCESS in 7m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/06ea3fca73d44ca988a5ec755a6b20ce : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6eaa328718194ad0baaac6d8384646e4 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db25c3b662f04d93a71b0150271de04f : SUCCESS in 9m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d74af940d8a4437a5511bb72f30aa9a : SUCCESS in 9m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7237fefe850148fea01a5586f6433d82 : SUCCESS in 1h 29m 02s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ae8ae79209d14c3da3c1b827a8b383c3 : FAILURE in 17m 17s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/9108692733964b8faf18cfdca5ec5b3c : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":3},{"id":"9f9040ebbb7c76662beb48313599c881ccb5897f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-02 15:31:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3d11d03aa7295b1708e4d98e9efe13e40d820a37","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-03 06:32:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c3ca2fe13678a4b3b97fe18f90ea7a05d473cd7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 06:32:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"23cf781474225b37a3a4508deb5a87940b1e7674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 07:54:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18b761d7d5c84cb6b649c8c352ebc0f8 : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37aac9b1b6e74f6f9f0507804e8ad196 : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b7d854608be49c488c1906a4188e599 : SUCCESS in 15m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e992b635ed045b6827675256043aef5 : SUCCESS in 10m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/21fd89b818764c78aa253f499ebbc608 : SUCCESS in 8m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c357d66b67f7449a85e0d1e97c3997db : SUCCESS in 12m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/978b034d3e8e4bb2b417857ba1f2faa5 : SUCCESS in 1h 19m 59s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/dcc4ef54d099476c8301faa6df60e86e : SUCCESS in 1h 16m 49s","accounts_in_message":[],"_revision_number":3},{"id":"7c7d186399c6eb33ecb730cc6fb6208182a637d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 07:54:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e247c20f81b89b7036022a03b56adbf1bcf311de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 07:56:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1986a2d01cf4a2d8ef7395a47c21f38 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"915304698628f717157344ae082eae383ee2d670","revisions":{"70058f6295a4d65af737c4699e4fa4c88c4fabd9":{"kind":"REWORK","_number":1,"created":"2020-05-26 15:24:36.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/53/730853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/53/730853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/1"}}},"commit":{"parents":[{"commit":"c5d52c16d26cd802c70e0fe983240ee406a2e256","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c5d52c16d26cd802c70e0fe983240ee406a2e256"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-05-26 15:22:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-05-26 15:22:22.000000000","tz":120},"subject":"Remove usage of six library","message":"Remove usage of six library\n\nNow we can move to python3 and forget the use of six library.\n\nChange-Id: If01ae1c8b16a4301506fa6cd1e893c4176a0aaa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/70058f6295a4d65af737c4699e4fa4c88c4fabd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/70058f6295a4d65af737c4699e4fa4c88c4fabd9"}]},"branch":"refs/heads/master"},"0b8e7e980729e86eba37949571c2e7f31fecc57b":{"kind":"REWORK","_number":2,"created":"2020-05-27 11:02:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/53/730853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/53/730853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/2"}}},"commit":{"parents":[{"commit":"c5d52c16d26cd802c70e0fe983240ee406a2e256","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c5d52c16d26cd802c70e0fe983240ee406a2e256"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-05-26 15:22:22.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-27 11:02:06.000000000","tz":0},"subject":"Remove usage of six library","message":"Remove usage of six library\n\nNow we can move to python3 and forget the use of six library.\n\nChange-Id: If01ae1c8b16a4301506fa6cd1e893c4176a0aaa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0b8e7e980729e86eba37949571c2e7f31fecc57b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0b8e7e980729e86eba37949571c2e7f31fecc57b"}]},"branch":"refs/heads/master"},"915304698628f717157344ae082eae383ee2d670":{"kind":"REWORK","_number":3,"created":"2020-06-02 11:39:39.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/53/730853/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/53/730853/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/53/730853/3"}}},"commit":{"parents":[{"commit":"c5d52c16d26cd802c70e0fe983240ee406a2e256","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c5d52c16d26cd802c70e0fe983240ee406a2e256"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-05-26 15:22:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-02 11:39:12.000000000","tz":120},"subject":"Remove usage of six library","message":"Remove usage of six library\n\nNow we can move to python3 and forget the use of six library.\n\nChange-Id: If01ae1c8b16a4301506fa6cd1e893c4176a0aaa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/915304698628f717157344ae082eae383ee2d670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/915304698628f717157344ae082eae383ee2d670"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
