)]}'
{"id":"openstack%2Fnetworking-bagpipe~731057","triplet_id":"openstack%2Fnetworking-bagpipe~master~I152c524375945fed8b5b687287b86eb1714ab87c","project":"openstack/networking-bagpipe","branch":"master","hashtags":[],"change_id":"I152c524375945fed8b5b687287b86eb1714ab87c","subject":"Remove usage of six","status":"ABANDONED","created":"2020-05-27 02:58:39.000000000","updated":"2020-06-01 06:16:14.000000000","total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"309c05a1e1f7801c9cee9ff0350748c8e900ebba","_number":731057,"virtual_id_number":731057,"owner":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-27 09:00:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":-1,"date":"2020-05-27 14:27:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-05-27 09:10:31.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2020-05-27 09:10:01.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"date":"2020-05-27 09:10:56.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2020-05-27 09:09:06.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-06-01 06:08:38.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"date":"2020-05-27 09:09:26.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"date":"2020-05-27 09:09:16.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 09:00:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-27 09:09:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-27 09:09:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-05-27 09:09:26.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-05-27 09:10:01.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-05-27 09:10:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-05-27 09:10:56.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-05-27 14:27:19.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-06-01 06:08:38.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"c12e116628a1baa3cfec14f88d85df0939198c23","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-05-27 02:58:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03f118c815844771d55524d69125e74fa3ee8cf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 04:28:58.000000000","message":"Patch Set 1: Verified-1\n\n(23 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e2027d228794aae84263ca25e872579 : FAILURE in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6456ee5861284a1690f40019767e325f : FAILURE in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2116c58f5cdc447ab4d3eec1b39c549d : FAILURE in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64bd2ddbf49a4ed591e180907a962716 : FAILURE in 15m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0137af497d6643d590ea0ac5d307320a : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b7c9972c9a54939a8bb93350d429ed2 : FAILURE in 9m 08s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd79b3ec67c14202b69271144d622f83 : FAILURE in 10m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbfb886f3ad9450cac17bc758210bbae : FAILURE in 10m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1cb9604a0c01457793f5861d78d3761a : FAILURE in 35m 11s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/9e1c3068749c4a2b97bef9248d2c3ae5 : FAILURE in 21m 23s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/65b83aef8a254bd587f7a8b27ee9a374 : SUCCESS in 1h 28m 31s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file networking_bagpipe/bagpipe_bgp/engine/tracker_worker.py has only 512 lines\u0027","accounts_in_message":[],"_revision_number":1},{"id":"ba2ed9f655a47e8871eb9e1af82ded585613f3a5","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-05-27 05:41:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92fe3a6eeba5d4f8ed553eca2c8e0c3398d91db6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 07:09:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c1dd3e5cf041e1ba2a308a9ee21340 : SUCCESS in 8m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed382dff08ff4aeea66b0a726679b3b8 : FAILURE in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dab4ac3607845229c45fa6e618282a6 : FAILURE in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd5cfb434954414a9e14c87211166abc : SUCCESS in 6m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bef005cda8c3497bbc6e04899bec2823 : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d11fb37eab046c1afab6bc4d10c0d7b : FAILURE in 6m 05s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a5a17237fd540c1b41fd5967dbe409e : FAILURE in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d36e64e30a90420493726a92d34c1d35 : FAILURE in 8m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/de28179896444f36ac8ecc7fc5273444 : SUCCESS in 1h 17m 37s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/54442750b8a148b895698f86ee8b4370 : FAILURE in 17m 20s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6cce440c219340978b12c1db9153b24c : SUCCESS in 1h 23m 11s","accounts_in_message":[],"_revision_number":2},{"id":"43f332dea1eb7eb61c86c2755d248f645fcfda9e","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-05-27 07:37:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9507ccbee756c2720365043ee5c7a7c0aaee6280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 09:00:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73c48f26a7447e1a659c5b5fb5dfc60 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/784df4f68bcd41b5a3409695276d8dba : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/062279b2ae2b450f9a81c478a5f6e39a : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2262acd8da5f48988d964d6113db94f0 : SUCCESS in 8m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/afb3f2a79ded4de9a03cd3471e85414a : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4098910691b8454eb96b99fc2f6a9f1a : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee6bcd8e14cb450c9bce003ae04abd22 : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6079312ae28a4799af1779e306d43146 : SUCCESS in 10m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e934330a5f024f51839887e512cf6748 : SUCCESS in 1h 20m 54s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f73bcdf06ce74c01aef4d499bace92e4 : FAILURE in 16m 25s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6cbe650963b54a329d21414786d3d199 : SUCCESS in 1h 15m 06s","accounts_in_message":[],"_revision_number":3},{"id":"c7312a5675ac4ebca6693ee0e9e39260e819df82","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-27 14:27:19.000000000","message":"Patch Set 3: Code-Review-1\n\nThere is a similar patch: https://review.opendev.org/730853","accounts_in_message":[],"_revision_number":3},{"id":"ab3c32aafb6ae125d698259622a1b95d62cbee66","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-01 06:08:38.000000000","message":"Patch Set 3:\n\nCould you abandon this? https://review.opendev.org/#/c/730853/ was proposed first.","accounts_in_message":[],"_revision_number":3},{"id":"309c05a1e1f7801c9cee9ff0350748c8e900ebba","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-06-01 06:16:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1ffae65b874b83beff07718ab5db98edb0a24168","revisions":{"53a68feab286edbe871936f4c26a444facab74b5":{"kind":"REWORK","_number":1,"created":"2020-05-27 02:58:39.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/57/731057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/57/731057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/1"}}},"commit":{"parents":[{"commit":"c5d52c16d26cd802c70e0fe983240ee406a2e256","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c5d52c16d26cd802c70e0fe983240ee406a2e256"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-05-26 20:54:49.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-05-26 21:12:42.000000000","tz":-240},"subject":"Remove usage of six","message":"Remove usage of six\n\nWith python3.x, classes can use \u0027metaclass\u003d\u0027 instead of\n\u0027six.add_metaclass\u0027, \u0027six.iteritems\u0027 and \u0027six.itervalues\u0027\ncan be replaced by \u0027items\u0027 and \u0027values\u0027, \u0027six.int2byte\u0027\ncan be replaced by \u0027bytes\u0027, \u0027six.moves.urllib\u0027 can be\nreplaced by \u0027urllib\u0027, \u0027six.moves.socketserver\u0027 can be\nreplaced by \u0027socketserver\u0027, \u0027six.class_types\u0027 can be\nreplaced by \u0027type\u0027, \u0027six.text_type\u0027 and \u0027six.string_type\u0027\nare just \u0027str\u0027.\n\nChange-Id: I152c524375945fed8b5b687287b86eb1714ab87c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/53a68feab286edbe871936f4c26a444facab74b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/53a68feab286edbe871936f4c26a444facab74b5"}]},"branch":"refs/heads/master"},"9a139377a7cd3c2458d96f2fc62dee0e82d78900":{"kind":"REWORK","_number":2,"created":"2020-05-27 05:41:32.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/57/731057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/57/731057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/2"}}},"commit":{"parents":[{"commit":"c5d52c16d26cd802c70e0fe983240ee406a2e256","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c5d52c16d26cd802c70e0fe983240ee406a2e256"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-05-26 20:54:49.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-05-26 23:55:27.000000000","tz":-240},"subject":"Remove usage of six","message":"Remove usage of six\n\nWith python3.x, classes can use \u0027metaclass\u003d\u0027 instead of\n\u0027six.add_metaclass\u0027, \u0027six.iteritems\u0027 and \u0027six.itervalues\u0027\ncan be replaced by \u0027items\u0027 and \u0027values\u0027, \u0027six.int2byte\u0027\ncan be replaced by \u0027bytes\u0027, \u0027six.moves.urllib\u0027 can be\nreplaced by \u0027urllib\u0027, \u0027six.moves.socketserver\u0027 can be\nreplaced by \u0027socketserver\u0027, \u0027six.class_types\u0027 can be\nreplaced by \u0027type\u0027, \u0027six.text_type\u0027 and \u0027six.string_type\u0027\nare just \u0027str\u0027.\n\nChange-Id: I152c524375945fed8b5b687287b86eb1714ab87c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9a139377a7cd3c2458d96f2fc62dee0e82d78900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9a139377a7cd3c2458d96f2fc62dee0e82d78900"}]},"branch":"refs/heads/master"},"1ffae65b874b83beff07718ab5db98edb0a24168":{"kind":"REWORK","_number":3,"created":"2020-05-27 07:37:58.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/57/731057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/57/731057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/57/731057/3"}}},"commit":{"parents":[{"commit":"c5d52c16d26cd802c70e0fe983240ee406a2e256","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c5d52c16d26cd802c70e0fe983240ee406a2e256"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-05-26 20:54:49.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2020-05-27 01:07:43.000000000","tz":-240},"subject":"Remove usage of six","message":"Remove usage of six\n\nWith python3.x, classes can use \u0027metaclass\u003d\u0027 instead of\n\u0027six.add_metaclass\u0027, \u0027six.iteritems\u0027 and \u0027six.itervalues\u0027\ncan be replaced by \u0027items\u0027 and \u0027values\u0027, \u0027six.int2byte\u0027\ncan be replaced by \u0027bytes\u0027, \u0027six.moves.urllib\u0027 can be\nreplaced by \u0027urllib\u0027, \u0027six.moves.socketserver\u0027 can be\nreplaced by \u0027socketserver\u0027, \u0027six.class_types\u0027 can be\nreplaced by \u0027type\u0027, \u0027six.text_type\u0027 and \u0027six.string_type\u0027\nare just \u0027str\u0027.\n\nChange-Id: I152c524375945fed8b5b687287b86eb1714ab87c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1ffae65b874b83beff07718ab5db98edb0a24168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1ffae65b874b83beff07718ab5db98edb0a24168"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
