)]}'
{"id":"openstack%2Fnetworking-bagpipe~733848","triplet_id":"openstack%2Fnetworking-bagpipe~master~I18b7d2f8ec64a591b6297cca292a4cd476b3a502","project":"openstack/networking-bagpipe","branch":"master","topic":"privsep","hashtags":[],"change_id":"I18b7d2f8ec64a591b6297cca292a4cd476b3a502","subject":"Add privsep boilerplate for bagpipe","status":"MERGED","created":"2020-06-05 14:02:11.000000000","updated":"2020-12-01 15:50:04.000000000","submitted":"2020-12-01 15:47:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733848-privsep","meta_rev_id":"bddce40989b55b38e45fb5e6a7914f53de8fbca4","_number":733848,"virtual_id_number":733848,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-12-01 12:38:18.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-01 15:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-11-13 08:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2020-12-01 08:45:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2020-12-01 08:45:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 18:35:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-11-09 16:11:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-13 08:53:07.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-12-01 08:45:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"347fd29a3df28780a35c654099bf3a4dc206dc70","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-05 14:02:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7dc4e2f0c1f883b5c7bf0d9d6503e91a6aeef6e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-05 14:02:39.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"08fa02be961f7a0630020f1849adbe4e361064e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 16:12:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/227e6c2d21524e668b2b777c7879d584 : FAILURE in 9m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8187b922a3cc435caeb350e12ca2a9ab : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/332b831ab1114f17869c1b50165580c6 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be1c28d9f14843aa8629a9751f3928d8 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8e0f943ecc64b0784a343ab418ae3ea : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35e7db71670b4a6eb1adb74b4aa67cc2 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53ad2088f1ff4ce3bc1a860d70e5b684 : SUCCESS in 9m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d233cd07c243138906acfecf168b5f : SUCCESS in 11m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/82d9b8cad1c542898e4424d30d880131 : TIMED_OUT in 2h 03m 10s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a15fa736e3fa45768fad842c44d80c3c : FAILURE in 21m 03s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0d7bf7cf8f61417ea179332f6f98a6ec : SUCCESS in 1h 33m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f55e8bb6110eac62d64f289c891a911a198cf73d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-08 12:30:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5bb357d5a5b4bb65980486baaf12b2188def057b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 14:38:20.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98ce8c0e55fa43e196a5a1765ba91a27 : FAILURE in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e05dbd46c0e4e18b4bd55efe530bad1 : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77616123deb44a578c7ecfc694f04611 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d4f6edee5b4dafba23bfbda22dfb50 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f08dc658d1da4da5bced5c6e04b5178e : SUCCESS in 3m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e3b17cf004bd4a8089ad80c9ee5bac55 : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/475a54218b38406b8ca49a0c1470bfff : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ef56a6e5e1e48659e2ac6ae758bfc62 : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5adefe907034ccdbe8cddfdccb82116 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d25f1674b68f4936aea35cf6ab5d8b3a : TIMED_OUT in 2h 03m 01s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/8ef6f0c378984964a76f963a315a140c : FAILURE in 18m 14s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/a348b6bd74e24cfe9f4edf8e2655c783 : SUCCESS in 1h 24m 04s","accounts_in_message":[],"_revision_number":2},{"id":"fc151a0e0270e2bce56d9780fc31ba6544e5648a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-08 17:35:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"40e1d1157aba73148eeb56f2f5f3cda815a5c8b7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-08 17:37:18.000000000","message":"Patch Set 2:\n\nI forgot that: please, add a task in https://storyboard.openstack.org/#!/story/2007686\n\nIf this project is not registered, then under \"openstack/governance\" project, with a name like:\n  \u0027networking-bagpipe - Migrate \"sysctl\" to oslo.privsep\u0027","accounts_in_message":[],"_revision_number":2},{"id":"a5c2ceac74c9652aed455683fb70aeed3f4811f4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-09 10:25:58.000000000","message":"Patch Set 2:\n\n\u003e I forgot that: please, add a task in https://storyboard.openstack.org/#!/story/2007686\n \u003e \n \u003e If this project is not registered, then under \"openstack/governance\"\n \u003e project, with a name like:\n \u003e \u0027networking-bagpipe - Migrate \"sysctl\" to oslo.privsep\u0027\n\nThanks Rodolfo, I check your comments","accounts_in_message":[],"_revision_number":2},{"id":"fa3141698f89b84a1062444bae6718914669dfb5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-09 15:35:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3e6ff707ab3a0b87616c8baed856331dfa1aeeec","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-09 15:46:16.000000000","message":"Patch Set 3:\n\n(5 comments)\n\nThanks for the help.\nI added task, do I need new task for every change here? Like new task for changing linux bridge filters, and mpls filters?","accounts_in_message":[],"_revision_number":3},{"id":"2447fb84886a78f338daf5ce36b777a79f6692b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 17:14:38.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c81331bd303547fca88b8a43d7b9f951 : FAILURE in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc1fe48b103e4fffa0eb07339d643077 : SUCCESS in 12m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c2c26c58faa48e6aca5eb200852e44c : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a863de621624df589e0c2aea5f7e567 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4318c750c383457d8c7afe4b8e7a6f15 : SUCCESS in 6m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/92ade2d6616545499097581002695110 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/136a90a570b14f02b120e5913431d15a : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d596b11993ad4eae99015c36f6692412 : SUCCESS in 9m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b990567354145a58b432d835297254e : SUCCESS in 9m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/17abe677f6ec4366b49cee55f0f9acdc : SUCCESS in 1h 07m 11s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2c966c41b38745ccb59faf9f11763932 : FAILURE in 22m 28s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/81de6272c1f64586bf70feaee9cb3371 : SUCCESS in 1h 30m 14s","accounts_in_message":[],"_revision_number":3},{"id":"ceaaa43bfb6e15cb43037e8f78dec41e23210608","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-11 14:24:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"60dbcf166e70c05a3251a597df2adf2ecc5d7752","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-11 14:54:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a374a24e0f18b1f439d423f3466be0b7ed85f799","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 16:00:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd026b66fdba49c59805a3976c77ec6d : FAILURE in 10m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18f0caa74212499b95ee63dfa81b2910 : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e6537b662aa40fe9dd8fd935544a49d : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ff08b3fa5b24864aa4d0a7ade198681 : FAILURE in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60c7c94da35144a3a368c83f976d8984 : SUCCESS in 5m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fc211e93de9d4fd8b241b74ec4976411 : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af65f04dea4840f3aa48e574114d9c56 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bee914026654dcaa99c6e1100f95fc5 : SUCCESS in 10m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e93b3ce1381e403abfafb95ab95ee973 : SUCCESS in 1h 27m 57s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a3b5c0773ee24e208d7084fbbe2a760e : FAILURE in 19m 11s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/262ba0c58f2d4a94b882fd110d6e472d : SUCCESS in 1h 35m 07s","accounts_in_message":[],"_revision_number":4},{"id":"0d8fb1bf2eb3e16bf66f5fa0aa1b51bf901c9462","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-16 12:18:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"93d0da9e4c7deb0ebebff9f73d0cf944644cff79","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-16 12:19:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d345b7fe73602e9c5786c8fec729f2f08109b64f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 13:36:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60c2dd3c50c9446bae72eee3c7f912df : SUCCESS in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b6ec76018c54b619b6578e52a183153 : SUCCESS in 12m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a6a7c2857f34ecba5820a4d406ad12f : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad51151633774989af5f82604bcf5df6 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4481a19552294d58973777576985b958 : SUCCESS in 6m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/62b1454608eb4a1d99ec962ee7290921 : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a4c2dd157eb4183b3e5c5a409722a9e : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/621177c02d824b3e89eb440ec52925d1 : SUCCESS in 12m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/15a2cca2335947f09cf03493b2b50a52 : SUCCESS in 1h 16m 17s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/86f0c0f116a7426e9fe6440ca2bb6cb6 : FAILURE in 21m 40s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/411d4ec2c74047819dfa8b3e36dc051c : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":5},{"id":"8043241926c83986352e066b3cbea15131d38b4d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-17 19:45:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"09e83b2b693221017e65ce8d83b7392b7e0353b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 21:30:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7403e40e9fcc4d2fa5e8ed97fee16558 : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd9d91d1f29140978eb3a3fd62308501 : SUCCESS in 11m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d21feea07f244eaae7dfe1dd570caa8 : SUCCESS in 10m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/349becc7cf044bd2b55005dc4fb02def : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a713f68be75421ead1a15c2cc11f91b : SUCCESS in 8m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fafc77d1371d49a2ae743026318d157e : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7caf48e3b344153a163011ea90b5802 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c044395ad8b45f8970f5dda79198479 : SUCCESS in 12m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/26247f6f6b3441fc802b887d97f83d70 : SUCCESS in 1h 27m 00s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/32a537b8959f475686412a3a5de25945 : FAILURE in 26m 18s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/08787ddfc7994f3bbd3989860400b828 : SUCCESS in 1h 22m 53s","accounts_in_message":[],"_revision_number":6},{"id":"624e32d19b102056f000e39270d6717860634f9d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-23 14:44:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"30cb4858635ac9bef4729ca9ad3cd536e0a07c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 16:18:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99b72d31acd49a1a161bd5998b7e031 : SUCCESS in 10m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8deecb3e7a9e4e97bb0a56525d542a5d : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/329f06793e0c4b018592b05b996a6f9c : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2140b9db213b4040ac2408a424f653b0 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aeff2e1683874dbaa468e76c97a9e016 : SUCCESS in 6m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9b1a8ba047ab4b1b9337282e8ac73a36 : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4ea0d2a180347888bacf1c035262f9a : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8f5a76e4b5041d390b232cc5d49577e : SUCCESS in 9m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/cfe3abf1680a452c8992b012cb38e6f9 : SUCCESS in 1h 27m 13s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/637904ffc5d84b75bc769fd295011919 : FAILURE in 20m 52s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/a3c2010a79084cb1b7ae2333990e44ff : SUCCESS in 1h 13m 18s","accounts_in_message":[],"_revision_number":7},{"id":"7d70ac5a476eb555404821ed3e9452417913a9a7","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-10-08 07:55:15.000000000","message":"Patch Set 7:\n\n(3 comments)\n\nSorry for taking soo long to look at this.\nThis is a nice addition.\n\nMy comments are only nits.","accounts_in_message":[],"_revision_number":7},{"id":"4138855ebd8b33c48ff8f8ca5285b8bd340f5302","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-12 09:54:21.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"4d602b8f25f4a7f48612fd6ea5d0e79b5884a9c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 11:30:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67e9e5b3bf724f0d9c957eeb5ff2e2bc : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a183c571a754243bd0a361afac8987f : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fcee1f8770241eeb485501307df63cc : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19df0b3ea36c4616bd20948b7d907aca : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79236fd21c6f4a2ba2b35cda7f3499b8 : SUCCESS in 5m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ba93c73528c458198b1ba44ca9da70f : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82a1d133b57e4c9a85a3860f86bf24f7 : FAILURE in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94a58e7c60464553a05d9cb1b44440e6 : SUCCESS in 8m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/cf273417549d49d8bffc641b51c1763b : FAILURE in 1h 24m 18s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/4216c06902f7464a83b398e3ec73b4ce : FAILURE in 16m 10s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/bf669546fe46455497981d569cb906f9 : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":8},{"id":"b2959146d37a7fb59e7a1cc2cebf981ab59acb93","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-12 14:39:02.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"66a1dcb9245ebbb933a88a4f6715f87c404f2ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 16:39:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e3a5e66c2f7495a83ef3c4c0891dd4d : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/859701a13bd54f2aa58d690234caab43 : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a201c75de0894cb0b0dee9a344561b10 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7e66716d4248f49f9bb0dad61a7ba5 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bd320c16c2c4f3eb86edd486d69e4d3 : SUCCESS in 4m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90d0e6fdcf09456d9be388125d8083cf : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b568a19b0f0d4432a85a919a75a1fce4 : FAILURE in 5m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d5949b7ee8240d1ab9e2dae8040b587 : SUCCESS in 8m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7d3bb3523c1e44b78e48b613ddebcf7a : SUCCESS in 1h 32m 27s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/49ce1478eebc455398df936ed1938142 : FAILURE in 15m 30s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/fd6825534bea437a96a6f89d77f9305c : SUCCESS in 1h 36m 40s","accounts_in_message":[],"_revision_number":8},{"id":"f92654e5a51d403669055810b4f92aa3afa1e38e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-12 17:11:11.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"856f6576b06ea56cd1883e0b37730e8415156671","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-12 17:12:41.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"18473c5f9c3dd8190f92c082b98ec7e1a49e8d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 19:22:56.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce94d6617c54573bbcbb32a6d540847 : SUCCESS in 10m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67527d5cd64c41f4a0d1123b8a8ec823 : SUCCESS in 8m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a7c76def31847d294c5b1bddd924c91 : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed818257c39c4a09805dda0393ce4826 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b9d20afcbbd42c99a4d560e7a16b7fe : SUCCESS in 4m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e41d2a31de0c4f80b4a89965724c0467 : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd892d1ab88d4f29a4834b1ac418ee35 : FAILURE in 8m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b83f514934204022b72ed09303c01d31 : SUCCESS in 13m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ec4b1445191f4467a21aa3deeacbe9ea : FAILURE in 1h 23m 48s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/67f0074fb8b0439494868d4c8814e393 : FAILURE in 20m 46s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/af3b6008479746e9bbdeafb5a1d11559 : SUCCESS in 1h 28m 49s","accounts_in_message":[],"_revision_number":9},{"id":"9bea00ad189e8df25254040b2b60f56f71f9478a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-13 09:14:33.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e365bb5c0c08ff4dbf94e5e55cda13208bb536f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 10:41:57.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc0984961d1246ac84eb83957d4b41db : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd4aca919db947fe8c6cd4594ea82ff9 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1bdc9fe3d67419a85c3335532bee9ab : SUCCESS in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86cdcc532184762a457db7608f3b523 : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a113b82fbae4b7eaaa921e7f0adab16 : SUCCESS in 5m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf4f752a82164edbbc991f0b01a36fe3 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e77bda38b2ad416d9d2a883fc3db5e11 : FAILURE in 4m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d3f30b382a34e3e984a9eda145c6ca2 : SUCCESS in 9m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/53592edb841148f5a9c7eb1957c7647e : FAILURE in 1h 20m 53s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/8d6d569815f943c98230a145e87e26f0 : FAILURE in 17m 58s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ddf7702cd32d40dcba92894ae452da8f : SUCCESS in 1h 14m 55s","accounts_in_message":[],"_revision_number":9},{"id":"21bae3e72f4918d06aec91f85322e8726737a7f1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-13 11:01:56.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"93a41344a15862e18bc39dd313f68d966281c462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 12:27:37.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fdd4562eed747c5817c4116e7356f55 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfafb89f11df45b382c3556af911f5a4 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44af58185b6049a2bfb481e526b6bf8d : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785bcfaa04c54011b1e1adb045ce051f : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd1e36465dbe4f0090b35fea2813bfe7 : SUCCESS in 4m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/91ac25c78c7545638f6beb6ff3a7bb15 : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/304c0d7af9554638916927ec4ea4288b : FAILURE in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f5570a4cbcb4f3283e4b80ce5b89c75 : SUCCESS in 8m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a34c8334f17b4fada19022f8b7cbcf2e : SUCCESS in 1h 19m 56s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/22ac173b51304010a3480d70c54197ab : FAILURE in 22m 33s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f86a47d782754a278b225c317a9cc5c2 : SUCCESS in 1h 16m 37s","accounts_in_message":[],"_revision_number":10},{"id":"227a63ab26d36ec3288e1d1387f4cbac34c1a87b","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-11-05 16:44:35.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"9a372b89040aa81931b35648fc882a53a1cb07e9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-05 18:35:43.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d75cc891dad1c96fd456b33c30a0422efd3bc370","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-09 12:13:28.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"428ed836bb2e1830b2c983d125fd187bda4f9c79","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-09 12:32:03.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"25d4b73cc247ce6c5d47b53fb1970defdf52b2bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 13:31:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cb0052e879d4a5e84e4c4d8b0122ccf : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e5ed5359b314d66b26c89f0d4e4c763 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/847ee36abb4c47f6968d4ba76bceec22 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc9308a7945040b7a1eb9e3c171b49db : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/885759d6135342b38c2d54aa814f07bf : SUCCESS in 30m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/030c66d4663444fb8af8107ab3ae9129 : SUCCESS in 8m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d7f9c7a328b34e1bb04a3a27e7022cad : SUCCESS in 6m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de732350b3834ae48639a05c4264537e : FAILURE in 4m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4a2d03df6814ee1b7e74ec48f4fd082 : SUCCESS in 8m 29s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/692727101be34e569d0b6d7ede0f467e : POST_FAILURE in 1h 12m 30s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/14b3177f3582472e8b6969d70ca9b54e : FAILURE in 15m 57s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/68ad83154f4944b4916e3fe52e5a1537 : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":11},{"id":"87d426616aded0c9b9a1fd573d1c16be749bc56f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-11-09 14:07:14.000000000","message":"Patch Set 11:\n\nrecheck\nPOST_FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"08167cf639bc5120688cd8f93b02d1b5a902f815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 16:11:16.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1adfc9c9b3344dd6957baa8fdc1d8080 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c95158d756fc491a981b078bcf876f47 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/817fa18197bd43aeaaa30dd0b2309208 : SUCCESS in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/810d0c847fd3416bb93bec6e59cfb872 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0b1ded874c44f7a801d9235e4b2abc : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22d381cc606645d3b40156e701b34cb4 : SUCCESS in 4m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a35c9e06cf84803b1db942a35ea4ee8 : SUCCESS in 3m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e440299911849379f26f62669f262d8 : FAILURE in 5m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4791d702fcd1408884da7628d6c60130 : SUCCESS in 8m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4a0e4d77ae8b4237bddd3aca817ac604 : SUCCESS in 1h 31m 54s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/868c1e377d6f4c59a357024b1f89c115 : FAILURE in 24m 54s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/2d4cec3e5a314800976271c9bbe7fd59 : SUCCESS in 1h 27m 11s","accounts_in_message":[],"_revision_number":11},{"id":"4df55b98b307f875fff0a7b5a4acf3c82a4598fa","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-11-13 08:52:51.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ba411a7b1c3a692afd66188b7e1ddef1b8c0b4f6","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-11-13 08:53:07.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"b2120a0eaeeaa5b704a92129f4eb80adadaa7929","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-01 08:45:09.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"eb1898f8e5cd291fadd68d6fb7c5e4e274185308","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 08:45:57.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"86a957e35a28175721adfa73d3024de25f23e4c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 10:09:25.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b5fcfc32e640d9a08dc2cb5ee4fe73 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/705cffff0ab547189becae4f21ef3e86 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9770216ec1c54861a4b6f4721a5dad98 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc1e9574ece412cb1a5cc36c44172ab : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90fced952710402a857822a979063df6 : SUCCESS in 3m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/03abba9afcd648d38a4f7f88351eb3a2 : SUCCESS in 2m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/fea38e754dd84b8193990aa5dd759593 : SUCCESS in 1h 17m 42s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/5d5b606ecd244e7585dbb368476d4772 : FAILURE in 1h 11m 56s","accounts_in_message":[],"_revision_number":11},{"id":"e0ce7e6ef438055ad0f6a12385d759a80e9def83","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-01 12:38:18.000000000","message":"Patch Set 11:\n\nrecheck\n\nvolume error (test_volume_create_get_update_delete_as_clone \u0026 test_volume_create_get_update_delete_from_image )","accounts_in_message":[],"_revision_number":11},{"id":"a968047574c81dee0cab42716bb784afb7954259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 14:05:04.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce60e3dba30c4700a53da262025277ce : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0be496b677774c1f846eed187d8c6079 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73deb96cbfcd4ddb95b22f4bd6353fb3 : SUCCESS in 8m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8fe5b053814438587a6b4aa71921371 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfd01cc448054605bc7c2a2acf902bdb : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c67cf4e193a45cb85ad74e6fd3b5740 : SUCCESS in 4m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/68f6df365d2d46e0b376a7e6291b8b58 : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27b5fbb86f0c47899a52cc831c6b2170 : FAILURE in 5m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b6b0e129f814d7898e9675aa97ade0e : SUCCESS in 8m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ee928c4a8545487cba87ee91b9eda2d9 : SUCCESS in 1h 19m 08s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0c420cc5a67740179135a3440ad2b0a1 : FAILURE in 18m 57s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/76bfc93c9aee423bb67874286d8e2d98 : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":11},{"id":"45b95de331efee9e8b019e303fc5f95ee79a67e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 14:07:35.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"988d23dc63f29e0f19e72b844cc97cdace398841","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 14:08:15.000000000","message":"Patch Set 11:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"50ee80dce0bd242dc279b0a43918bc6ad8f1efb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 14:09:06.000000000","message":"Patch Set 11:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"19dd968ac068dbd73a1be6f23ed062c8e65ba680","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:47:31.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de8e81110a245a198a3bdcbf51a5fa9 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c857db53a7a4ea7b17e844d18885fcf : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26cbc633ccff4cb0bde1224ea27a1286 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7bf553ecebd40ebac39d447c38d40de : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dc3cda7c3764f718ff53099fd065024 : SUCCESS in 4m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/02d5932ff9ed478393af46d6fc3b0339 : SUCCESS in 2m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ed1327ea33ea49bf8a8f5cffa211260c : SUCCESS in 1h 14m 38s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/19411c275ddf4a70aacd57de6d3c6125 : SUCCESS in 1h 24m 10s","accounts_in_message":[],"_revision_number":11},{"id":"176843288554e7fa6dbd834b05fff3cd63201abb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:47:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"bddce40989b55b38e45fb5e6a7914f53de8fbca4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:50:04.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d4dd5d6b2d4f0ba6270832a2d69bfa : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1785fa8a47c4649a208b0fab5ce3fac : SUCCESS in 55s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"93d660c210371827cff4a4f5d944fbb97d462a9d","revisions":{"dfeb8cc183d20aa2f66c137c4627b1e6026372e8":{"kind":"REWORK","_number":1,"created":"2020-06-05 14:02:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/1"}}},"commit":{"parents":[{"commit":"915304698628f717157344ae082eae383ee2d670","subject":"Remove usage of six library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/915304698628f717157344ae082eae383ee2d670"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dfeb8cc183d20aa2f66c137c4627b1e6026372e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dfeb8cc183d20aa2f66c137c4627b1e6026372e8"}]},"branch":"refs/heads/master"},"887b9e840bc21b71997adc5c4445d52459dd7f17":{"kind":"REWORK","_number":2,"created":"2020-06-08 12:30:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/2"}}},"commit":{"parents":[{"commit":"915304698628f717157344ae082eae383ee2d670","subject":"Remove usage of six library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/915304698628f717157344ae082eae383ee2d670"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-08 12:29:51.000000000","tz":120},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/887b9e840bc21b71997adc5c4445d52459dd7f17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/887b9e840bc21b71997adc5c4445d52459dd7f17"}]},"branch":"refs/heads/master"},"f1dfbce33664d7a850888e696c1115f778e7ed03":{"kind":"REWORK","_number":3,"created":"2020-06-09 15:35:28.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/3"}}},"commit":{"parents":[{"commit":"915304698628f717157344ae082eae383ee2d670","subject":"Remove usage of six library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/915304698628f717157344ae082eae383ee2d670"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-09 15:31:42.000000000","tz":120},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\nStory: #2007686\nTask: #40027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1dfbce33664d7a850888e696c1115f778e7ed03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f1dfbce33664d7a850888e696c1115f778e7ed03"}]},"branch":"refs/heads/master"},"7aba02088954f1a6593735ae5942145d0c3d006e":{"kind":"REWORK","_number":4,"created":"2020-06-11 14:24:08.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/4"}}},"commit":{"parents":[{"commit":"915304698628f717157344ae082eae383ee2d670","subject":"Remove usage of six library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/915304698628f717157344ae082eae383ee2d670"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-11 14:23:45.000000000","tz":120},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\nStory: #2007686\nTask: #40027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7aba02088954f1a6593735ae5942145d0c3d006e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7aba02088954f1a6593735ae5942145d0c3d006e"}]},"branch":"refs/heads/master"},"733733a9aaccf8da70f85ee2a01505e3849703a3":{"kind":"REWORK","_number":5,"created":"2020-06-16 12:18:07.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/5"}}},"commit":{"parents":[{"commit":"915304698628f717157344ae082eae383ee2d670","subject":"Remove usage of six library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/915304698628f717157344ae082eae383ee2d670"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-16 12:11:24.000000000","tz":120},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\nStory: #2007686\nTask: #40027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/733733a9aaccf8da70f85ee2a01505e3849703a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/733733a9aaccf8da70f85ee2a01505e3849703a3"}]},"branch":"refs/heads/master"},"1065a8d7c3e8ade523fd5cbd73e2cdfbb0ece247":{"kind":"REWORK","_number":6,"created":"2020-06-17 19:45:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/6"}}},"commit":{"parents":[{"commit":"915304698628f717157344ae082eae383ee2d670","subject":"Remove usage of six library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/915304698628f717157344ae082eae383ee2d670"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-17 18:53:38.000000000","tz":120},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\nStory: #2007686\nTask: #40027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1065a8d7c3e8ade523fd5cbd73e2cdfbb0ece247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1065a8d7c3e8ade523fd5cbd73e2cdfbb0ece247"}]},"branch":"refs/heads/master"},"95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0":{"kind":"REWORK","_number":7,"created":"2020-06-23 14:44:38.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/7"}}},"commit":{"parents":[{"commit":"915304698628f717157344ae082eae383ee2d670","subject":"Remove usage of six library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/915304698628f717157344ae082eae383ee2d670"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-23 14:42:13.000000000","tz":120},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\nStory: #2007686\nTask: #40027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0"}]},"branch":"refs/heads/master"},"4ee5dff5161153a8540359dd3993db5e17de24b0":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-10-12 09:54:21.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/8"}}},"commit":{"parents":[{"commit":"7e96cc651394813c1dc80747186b6cfcaa173f14","subject":"Merge \"Add Python3 wallaby unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7e96cc651394813c1dc80747186b6cfcaa173f14"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-10-12 09:54:21.000000000","tz":0},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\nStory: #2007686\nTask: #40027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4ee5dff5161153a8540359dd3993db5e17de24b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4ee5dff5161153a8540359dd3993db5e17de24b0"}]},"branch":"refs/heads/master"},"00674f465b4b0ad45507f3decf141a3e7bf4c467":{"kind":"REWORK","_number":9,"created":"2020-10-12 17:11:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/9"}}},"commit":{"parents":[{"commit":"7e96cc651394813c1dc80747186b6cfcaa173f14","subject":"Merge \"Add Python3 wallaby unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7e96cc651394813c1dc80747186b6cfcaa173f14"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-10-12 17:10:53.000000000","tz":120},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\nStory: #2007686\nTask: #40027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/00674f465b4b0ad45507f3decf141a3e7bf4c467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/00674f465b4b0ad45507f3decf141a3e7bf4c467"}]},"branch":"refs/heads/master"},"9dba51bfcad0a5a9000f2c1de46ed5f9434aca35":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2020-10-13 11:01:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/10"}}},"commit":{"parents":[{"commit":"7e96cc651394813c1dc80747186b6cfcaa173f14","subject":"Merge \"Add Python3 wallaby unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7e96cc651394813c1dc80747186b6cfcaa173f14"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-10-13 11:01:52.000000000","tz":0},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\nDepends-On: https://review.opendev.org/755719\nStory: #2007686\nTask: #40027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9dba51bfcad0a5a9000f2c1de46ed5f9434aca35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9dba51bfcad0a5a9000f2c1de46ed5f9434aca35"}]},"branch":"refs/heads/master"},"93d660c210371827cff4a4f5d944fbb97d462a9d":{"kind":"REWORK","_number":11,"created":"2020-11-09 12:13:28.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/733848/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/733848/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/733848/11"}}},"commit":{"parents":[{"commit":"7e96cc651394813c1dc80747186b6cfcaa173f14","subject":"Merge \"Add Python3 wallaby unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7e96cc651394813c1dc80747186b6cfcaa173f14"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-05 14:00:22.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-11-09 12:12:57.000000000","tz":60},"subject":"Add privsep boilerplate for bagpipe","message":"Add privsep boilerplate for bagpipe\n\nAdd boilerplate for using privsep in networking-bagpipe and change\ncalling sysctl to use privsep.\n\nChange-Id: I18b7d2f8ec64a591b6297cca292a4cd476b3a502\nDepends-On: https://review.opendev.org/755719\nStory: #2007686\nTask: #40027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/93d660c210371827cff4a4f5d944fbb97d462a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/93d660c210371827cff4a4f5d944fbb97d462a9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
