)]}'
{"id":"openstack%2Fnetworking-bagpipe~737554","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ifedff9e638a077f3ea092fc82d48db12c777c816","project":"openstack/networking-bagpipe","branch":"master","topic":"privsep","hashtags":[],"change_id":"Ifedff9e638a077f3ea092fc82d48db12c777c816","subject":"Implement \"modprobe\" using oslo.privsep","status":"MERGED","created":"2020-06-23 14:44:46.000000000","updated":"2020-12-01 22:14:33.000000000","submitted":"2020-12-01 22:13:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737554-privsep","meta_rev_id":"468d1f92d5bd353b3b02c47003dc97d53c7b53a9","_number":737554,"virtual_id_number":737554,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-01 22:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-12-01 22:14:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-12-01 20:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2020-12-01 22:14:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-12-01 20:36:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 08:47:14.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-10-13 09:38:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-10-13 10:48:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-19 21:15:01.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"86427e544ca45f71ad78910ad657d02bcd0bb249","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-23 14:44:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f91fed0bcf4c4a6498ca1acea3140865e11f83c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 16:10:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f0c64d20da147cb95c2581ec285748d : SUCCESS in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ae435561470466798344ebcdc046939 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06d393e4941d493e9d024f33456ded9e : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d258ddfa88214d7cbcca1fa802564f5b : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a257f115e2c43efaca209572bfddf24 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca949f643bb449e3a30d2402836a1c9a : SUCCESS in 8m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3478d6b6d53040eb8db74475ff3c990a : SUCCESS in 1h 21m 40s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/91fbff72824d4197bd9c3a3592adb573 : FAILURE in 19m 10s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/90124f75e7c24eb3baf35a04751e70d4 : SUCCESS in 1h 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"07676fbbce8dbd61e87bb194db9fbf90a11b0b83","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-06-24 08:21:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b1930cf6389097acdea83b5b50c3e086ba0e6583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 09:44:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02be2cb1432c46c692dbea9d68aadd9b : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30c1b0f22e424460b20f22611b1f128a : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d0e450717594efa81cb1117498b7c60 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c989308fe52a43e8944de20f115772de : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/355cda80c1fd45e2843cd7b6152ecb2c : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52709852359847989aedae4abfe9631e : SUCCESS in 10m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5a0d723e4b01424b863fefd2f014893c : SUCCESS in 1h 14m 14s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/50d3b0f411d0457eacc8ced17decc255 : FAILURE in 14m 02s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f1b1e0fb39ca418b8f0c0ef97c8e1f39 : SUCCESS in 1h 18m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f6a37ae5aa620b825d695f4c2d54f2aa2b43c69c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-01 09:54:23.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6e223eefb266d66ae29fd2261c524d53f53eccbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 11:43:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc0672ad30b4b0d8e76f3528d437580 : SUCCESS in 10m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8724370cadb469588962de629830cc6 : SUCCESS in 11m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/485cac22509f4892be49df9fead9ebe4 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec2f6a1da0c4d6ab94668b3519f8f4b : SUCCESS in 13m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d77446bc9cb5482bbcd9bf1b88cf4abc : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82b5760b435e47ae9a1c892d70ea8c2d : SUCCESS in 10m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f9b4a5bfa9a349efae377a4e66b1a62f : SUCCESS in 1h 30m 05s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/25bef1ffaf1446d5a91e0cdfe23e8abb : FAILURE in 22m 07s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1dc7ba356e3042f5afa77138db377cc5 : SUCCESS in 1h 38m 20s","accounts_in_message":[],"_revision_number":3},{"id":"6de21d758e5c63d2cabdcee224c1879bbea0e11c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-01 15:44:06.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4daed9c8883311725d5ef6c40855321dbb745bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 17:56:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee5cc610328640f6875d5c7b6e11c4a8 : SUCCESS in 12m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55fe3ed52c1541868b1f8e9cc84bf828 : SUCCESS in 12m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4c2451b21854adf81f7e76415b6b088 : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d5e53373be4ebe86d595d8b7c6e0ed : SUCCESS in 18m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32267453b4f14289a27ad49214f98e47 : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12479f3d07da4279942f3b36da0ca1b7 : SUCCESS in 12m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f5971bbde959464b90cf4a5fdeee5f46 : SUCCESS in 1h 51m 15s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/1ba4b278f29d43e09d63e9f81be385a0 : FAILURE in 26m 37s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/5d44d8096e6f4afe9667091eb52036ee : POST_FAILURE in 2h 08m 55s","accounts_in_message":[],"_revision_number":4},{"id":"4c2fa3ec09c92ae3b5aa3496265bb06280ff110e","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-10-08 08:46:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f5ca27f6dbfb85567721c495e943b4b5e8b1c39d","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-10-08 08:47:14.000000000","message":"Patch Set 4:\n\nThis looks good to me.","accounts_in_message":[],"_revision_number":4},{"id":"b1aaab680b9317bc3a76fdcc67bfe8d35894799a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-08 10:12:19.000000000","message":"Patch Set 4:\n\n\u003e This looks good to me.\n\nThanks Thomas, if you have any feedback regarding scaling issues or similar with it please share with me, as I have only some local devsatck env during the desing.\nBy the way this will be this cycle\u0027s community goal: https://governance.openstack.org/tc/goals/proposed/migrate-to-privsep.html","accounts_in_message":[],"_revision_number":4},{"id":"db14d5964a4a2700b360207c95c5ccce2c0a5522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 10:27:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b1a52886a0841e489e46095589da801 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76a7ab51604c4f69b94818502cbfceab : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22e5d5c3709a4794a4c2faa810e1d628 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fff82d073674947a0e6b43b0f5dbd3d : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85da5138e13e4ec1acf6bade203f9822 : FAILURE in 5m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50b8600eb68543c7907bd0fef7bdf13f : SUCCESS in 8m 53s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f0975e839ecb46459f3a3ccca3cc8920 : SUCCESS in 1h 32m 36s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/76ca923e82ad48e988f2e25a7a64fb1f : FAILURE in 21m 23s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ac17447e875b48cba36aee7645843cf6 : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":4},{"id":"2be50b503bfe56304a71697881a116cdfe40b795","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-12 09:54:41.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"4b458b8ff76a7cf6c63da2166970102e03581289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 12:59:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6912f7a5089478db8d0187dd9d90e2e : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09ef2ad44c5c4ec38e742b188d5c32c5 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39d06e926275420e861b40b57f79f3d4 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a303d56c233441bb85e7dc4ba15becce : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d39cb61c66d41728c035795aa928fb1 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd67e3a7b4e44c91bf1f60566bfbb8a4 : SUCCESS in 8m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1b735470842d4b69868368470b15271a : SUCCESS in 1h 18m 58s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/dfc65f0dc31942ab92ff48dc2c5fcf9a : FAILURE in 18m 13s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/8df46cbd78204102ba4eacb051865b13 : SUCCESS in 2h 45m 25s","accounts_in_message":[],"_revision_number":5},{"id":"8e03d38418d1ab291ad800377b38d63d4fb35422","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-13 09:14:44.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"e9b3513f854c830675937c314fe8a1999ebfe68e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-13 09:38:39.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"929e3ae49e8a43697133a89302623c528eeca43b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 10:48:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d168483367c84e8b99523634127d9d58 : SUCCESS in 9m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b4edc94e6ec4bcd89cabd568b2df254 : SUCCESS in 9m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/264c18e639e646819ea5a441a1f71185 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff65d78fce274bed901249d734023d61 : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a60f3259e9a9490497d582eb083362b7 : FAILURE in 4m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0b66de979fb4aa392711a34d0417779 : SUCCESS in 8m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5495b0c546bc4e9aabe764319b7cfb28 : SUCCESS in 1h 27m 11s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/5205d19959d840e8a355b2a2442a09cb : FAILURE in 24m 11s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/aced6622ceab4e34bfdb9581355a0b1e : SUCCESS in 1h 19m 50s","accounts_in_message":[],"_revision_number":6},{"id":"a13ac230c4352246d677644ae9e045dfbe27cd1a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-19 21:15:01.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"343e8fdbc0da7bbafdb25094b84b1d0f72b4a819","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-01 18:46:17.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"91777f2d4dff7d8d24a3a8bed83e91f6ab289ef6","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-12-01 20:36:19.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"d8f32a9a1e414cb1931c4f7f20875d4dc0aa5c33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 20:41:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d7b572fff146af9d4822755e25e3f0 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8567821d2a9c45588480f52424f10b18 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21449dc8404f49d9982e7fe2b8f7f49d : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/080b9165571147ccb0e8a127b9be665a : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17597c7ddb2d4f90a41785ba9580b4b3 : SUCCESS in 7m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12287a8138bb491da022f194e7471138 : FAILURE in 21m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5db3798e77eb4221a4308e95b79509fc : SUCCESS in 7m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7a71ad27551a4a37a4fd8907ddd2b091 : SUCCESS in 1h 22m 16s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e0f43f550e4d41b697521e3e66586dd1 : FAILURE in 15m 44s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/95220ef381ca4057b493e29d9355a247 : SUCCESS in 1h 46m 29s","accounts_in_message":[],"_revision_number":7},{"id":"170b68e67d22172f577b8378ef2d53f3db2ab66a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 20:41:56.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"09bf834ec51d7172cadadb5b5684c7a5c0b44a41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 22:13:07.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94605c4e410e4f3395ffc13836e4fec4 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8c9037e8e5a4715b2c06d97df086972 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd778f2469eb4f8bbc2f91a131ca1dff : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbbc6f5184194dceb5ecc72919236ac8 : SUCCESS in 6m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a3d595c8eb794f12b15aef05d282996b : SUCCESS in 1h 19m 29s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3942608cd48b4ff2ac83757f773f4c16 : SUCCESS in 1h 24m 28s","accounts_in_message":[],"_revision_number":7},{"id":"2287b27fced06ae5efc841388d5d769447db1291","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 22:13:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"2e0032049a48fafd3fe5e35e3703739074e814e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 22:14:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d2a36a880dd428fa4db656dd092dd7e : SUCCESS in 58s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"fbda54e1601773f762439f7ce7c70ff97a5484b6","revisions":{"fba2fdce87d92dea03b0fc08de7f2d18c47a11df":{"kind":"REWORK","_number":1,"created":"2020-06-23 14:44:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/54/737554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/54/737554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/1"}}},"commit":{"parents":[{"commit":"95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0","subject":"Add privsep boilerplate for bagpipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-23 11:18:37.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-23 14:43:09.000000000","tz":120},"subject":"Implement \"modprobe\" using oslo.privsep","message":"Implement \"modprobe\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"modprobe\" commands\nand replace them with oslo.privsep.\n\nStory: #2007686\nTask: #39974\nChange-Id: Ifedff9e638a077f3ea092fc82d48db12c777c816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fba2fdce87d92dea03b0fc08de7f2d18c47a11df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fba2fdce87d92dea03b0fc08de7f2d18c47a11df"}]},"branch":"refs/heads/master"},"c9a93d9282acbec8966b0a0dfabae71689e8959c":{"kind":"REWORK","_number":2,"created":"2020-06-24 08:21:52.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/54/737554/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/54/737554/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/2"}}},"commit":{"parents":[{"commit":"95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0","subject":"Add privsep boilerplate for bagpipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-23 11:18:37.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-24 08:21:28.000000000","tz":120},"subject":"Implement \"modprobe\" using oslo.privsep","message":"Implement \"modprobe\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"modprobe\" commands\nand replace them with oslo.privsep.\n\nStory: #2007686\nTask: #39974\nChange-Id: Ifedff9e638a077f3ea092fc82d48db12c777c816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c9a93d9282acbec8966b0a0dfabae71689e8959c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c9a93d9282acbec8966b0a0dfabae71689e8959c"}]},"branch":"refs/heads/master"},"6756b247e2f0bbbc4c4052e80690c3bcb0af3440":{"kind":"REWORK","_number":3,"created":"2020-07-01 09:54:23.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/54/737554/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/54/737554/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/3"}}},"commit":{"parents":[{"commit":"95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0","subject":"Add privsep boilerplate for bagpipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-23 11:18:37.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-07-01 09:54:19.000000000","tz":0},"subject":"Implement \"modprobe\" using oslo.privsep","message":"Implement \"modprobe\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"modprobe\" commands\nand replace them with oslo.privsep.\n\nStory: #2007686\nTask: #39974\nChange-Id: Ifedff9e638a077f3ea092fc82d48db12c777c816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6756b247e2f0bbbc4c4052e80690c3bcb0af3440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6756b247e2f0bbbc4c4052e80690c3bcb0af3440"}]},"branch":"refs/heads/master"},"34a05bb41183f7a11d1e5453e073c0406dd138ce":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-07-01 15:44:06.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/54/737554/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/54/737554/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/4"}}},"commit":{"parents":[{"commit":"95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0","subject":"Add privsep boilerplate for bagpipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/95fb6ebdd3c4edbe0e230a18682ab2f9f5e62bf0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-23 11:18:37.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-01 15:42:58.000000000","tz":120},"subject":"Implement \"modprobe\" using oslo.privsep","message":"Implement \"modprobe\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"modprobe\" commands\nand replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40259\nChange-Id: Ifedff9e638a077f3ea092fc82d48db12c777c816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/34a05bb41183f7a11d1e5453e073c0406dd138ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/34a05bb41183f7a11d1e5453e073c0406dd138ce"}]},"branch":"refs/heads/master"},"95baacc856933d3c71bfc6b0cfcbcb3777ecc6dc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-10-12 09:54:41.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/54/737554/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/54/737554/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/5"}}},"commit":{"parents":[{"commit":"4ee5dff5161153a8540359dd3993db5e17de24b0","subject":"Add privsep boilerplate for bagpipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4ee5dff5161153a8540359dd3993db5e17de24b0"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-23 11:18:37.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-10-12 09:54:41.000000000","tz":0},"subject":"Implement \"modprobe\" using oslo.privsep","message":"Implement \"modprobe\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"modprobe\" commands\nand replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40259\nChange-Id: Ifedff9e638a077f3ea092fc82d48db12c777c816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/95baacc856933d3c71bfc6b0cfcbcb3777ecc6dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/95baacc856933d3c71bfc6b0cfcbcb3777ecc6dc"}]},"branch":"refs/heads/master"},"3c32b95b527ddaaaaebc1a8bb381b846283519a5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-10-13 09:14:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/54/737554/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/54/737554/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/6"}}},"commit":{"parents":[{"commit":"00674f465b4b0ad45507f3decf141a3e7bf4c467","subject":"Add privsep boilerplate for bagpipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/00674f465b4b0ad45507f3decf141a3e7bf4c467"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-23 11:18:37.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-10-13 09:14:44.000000000","tz":0},"subject":"Implement \"modprobe\" using oslo.privsep","message":"Implement \"modprobe\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"modprobe\" commands\nand replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40259\nChange-Id: Ifedff9e638a077f3ea092fc82d48db12c777c816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3c32b95b527ddaaaaebc1a8bb381b846283519a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3c32b95b527ddaaaaebc1a8bb381b846283519a5"}]},"branch":"refs/heads/master"},"fbda54e1601773f762439f7ce7c70ff97a5484b6":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-12-01 18:46:17.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/54/737554/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/54/737554/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/54/737554/7"}}},"commit":{"parents":[{"commit":"93d660c210371827cff4a4f5d944fbb97d462a9d","subject":"Add privsep boilerplate for bagpipe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/93d660c210371827cff4a4f5d944fbb97d462a9d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-06-23 11:18:37.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-12-01 18:45:05.000000000","tz":60},"subject":"Implement \"modprobe\" using oslo.privsep","message":"Implement \"modprobe\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"modprobe\" commands\nand replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40259\nChange-Id: Ifedff9e638a077f3ea092fc82d48db12c777c816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fbda54e1601773f762439f7ce7c70ff97a5484b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fbda54e1601773f762439f7ce7c70ff97a5484b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
