)]}'
{"id":"openstack%2Fnetworking-bagpipe~738872","triplet_id":"openstack%2Fnetworking-bagpipe~master~I21e1d5f1e446afffaba6002f277a2990f934de29","project":"openstack/networking-bagpipe","branch":"master","topic":"privsep","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-04-17 15:13:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I21e1d5f1e446afffaba6002f277a2990f934de29","subject":"Implement \"brctl\" and \"bridge\" using oslo.privsep","status":"MERGED","created":"2020-07-01 15:44:06.000000000","updated":"2023-04-17 15:14:08.000000000","submitted":"2023-04-17 15:13:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"738872-privsep","meta_rev_id":"5df3c5187a85dbfd61d291dfaac21087ffc1eb16","_number":738872,"virtual_id_number":738872,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-17 15:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-17 13:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-04-17 13:43:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 08:45:16.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-10-13 09:36:34.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-10-13 10:42:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 11:09:05.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"f0b715fcf0c549aea5d95cd8ef6adacdcce640ca","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-01 15:44:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6d785bd8c90839f25a0499d40e3e5b22c99293e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 17:12:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe6ede5025742f38421c2629d078525 : SUCCESS in 10m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9cae4bfb1d54b588e345e196aecf0b0 : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8c8b01ab2ea4e4a94f756d341495390 : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58614da10ba4bec84fd14717906e1f1 : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf8c4d0f7ae0493c88fe98201c832ea0 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39d210b4b55d4ad38177363e60ad8333 : SUCCESS in 9m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/58841f8993df4220ba4ae16d78e628ea : SUCCESS in 1h 23m 56s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/d7927ecf86e34ae78529b6196937e78c : FAILURE in 21m 46s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1dfd9fe91e1b4f3585ee1cac50e0bc79 : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":1},{"id":"54b8021ada2059c1cfaaad2af72a0fceeea51367","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-10-08 08:45:16.000000000","message":"Patch Set 1: Code-Review+2\n\nThis looks ok.\n\nOne side comment: I wonder if it wouldn\u0027t be even nicer to rely on pyroute2 instead of using the CLI here. I would tend to think that it should have the functionality required.","accounts_in_message":[],"_revision_number":1},{"id":"8405a843b7ba6550524ea7f2f4f975087a6c5b73","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-12 09:54:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3c78f92f6d81bcb6644ff1e4fd96d81e5c47f621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 12:54:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/094836cd7c2a4868b5b03c0334cfc321 : SUCCESS in 10m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba9285da81814f2b9a9b0cc0812f6200 : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5035a88ce1de49048069b3d34c62c957 : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec2eb529c574e639fdea4ae33b914d2 : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ca7f02fbb8d494f9393e4d2575c3cda : FAILURE in 5m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/776247dd4ec149e49229f1f21c81cfa9 : SUCCESS in 10m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/54e94a9ae96742268e86e7a9d12c269d : SUCCESS in 1h 30m 38s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/59dadb0f21214de2b67d36b56e24090f : FAILURE in 19m 22s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/08cf718d00d04a8b8724cf512f0f918f : SUCCESS in 1h 16m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d9aacef9bf37a33879e741a4e6baa785b2af78a5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-13 09:14:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"724a153d21c230a8f93d8404f85bfc2ffaaf1564","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-10-13 09:36:34.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a96b550e4af3dfc97623e04ecbb5d30df254185c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 10:42:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/598cdc04727d479bb208439d3de7f2ce : SUCCESS in 8m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/643e05be4f8d4473b409da59c0d26431 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e6a8a5364034671b0d24bee2ff14081 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca2d48152aa4b4c8c8803439e607e48 : SUCCESS in 9m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41fac09b7a39438aa38371cd33152edc : FAILURE in 5m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d631e04a8f448838e6ca406af2c4a1f : SUCCESS in 8m 21s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/883389b3b30d4e8988bfe054deebc8fb : FAILURE in 1h 17m 56s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/303fe15a2eb34f62bd907db48ffef833 : FAILURE in 25m 59s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ee7b2dd0c4744675834ae0f13ca9eb16 : SUCCESS in 1h 20m 01s","accounts_in_message":[],"_revision_number":3},{"id":"7981cf7382916e41bb73c2e123f66e1eaf5486c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-01 18:48:29.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c4dcee2809a07221175eeea2b5946512742e5f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 20:23:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce1e52d7d7e24c39960966401f3b8552 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a7894d3f6e84916b8727c6d93734ed6 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f34beb0137a4f89bf1957b11080a86b : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdb35668785d4fb39aa7db0a7296a93b : SUCCESS in 25m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87ae43a9c3d0478a887fe30085c5e0d2 : TIMED_OUT in 31m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0289e7ce25cc40d2b17761232dd5b5af : FAILURE in 6m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35215055c78b451aba02bc7913d040a7 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/694fd8c616af46eeb9af35ac3901f9ec : SUCCESS in 1h 20m 43s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/827d46f7a29f482bb6e0c23b097c154f : FAILURE in 22m 12s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/da9d811e805646cbba9af8275df3f51c : SUCCESS in 1h 25m 32s","accounts_in_message":[],"_revision_number":4},{"id":"7a8e74fbd8a8dd46bd171f44751f1dc53d879b52","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-12-02 09:08:15.000000000","message":"Patch Set 4:\n\nrecheck\ndoc build timeout","accounts_in_message":[],"_revision_number":4},{"id":"5ac388a18017dcf8225caaf44d131211610efb2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 10:33:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91eb37a604234a598cd5c6a3c25795d9 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/925e4dbb5c8b4e1a8c60d37643870185 : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e86b0453baca464187199e8529ee1953 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e0ca2f7bc21403fbeb66f899e0721fa : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a936ff95a1240298841e00d41e4f717 : SUCCESS in 6m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8b33999dc924e869a065e720448cb3a : FAILURE in 4m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f97b13355c9d47d9963015c626e5b131 : SUCCESS in 8m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1b28d07230614e89ac9b4b05af23da96 : SUCCESS in 1h 18m 20s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/68de4d87cc88458c804ad6a1031324bf : FAILURE in 14m 43s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c891a1df4b14434d9db931aa4427cc37 : SUCCESS in 1h 22m 17s","accounts_in_message":[],"_revision_number":4},{"id":"fa9067aa90abaa895f294d1e1e23676ec02e5b86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-10 07:34:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bce4381f6bb50b7e20d3bb92bf5aaa27e768e5f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-10 07:39:13.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"fac40f6babf031aaa4545f8a604406fa5cbd97cd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-10 07:41:13.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"73649091a0bd3d3fe2d7184d876f0f7ef539b44f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 09:03:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a132d87a7e4c0b98c8f81e75f31ce1 : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32bd4b21219f49cc8b068418d3f99b2f : SUCCESS in 10m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f15a6ecebf3146159b25c79a7add8828 : SUCCESS in 9m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c959fc63605e4ef28aeffbf8ee327230 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1c999da58f4b78be6f23bd5f4e37a0 : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81b96002a59c459ca07c7ce1fe295e90 : FAILURE in 4m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb0dcccb9657401b93a5761bdc4f4749 : SUCCESS in 10m 36s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/265a5b27c628493ea895b43e8e9d2462 : SUCCESS in 1h 11m 03s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a20602004d174abd9c893ba5192e01ef : FAILURE in 15m 49s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b75cd7426507475c9c641fa55f6d37e4 : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":6},{"id":"0881c554a2591322601ac3d175b49d053cccf84b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-11 12:04:27.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9d51c7d2f97b8bd13d13d0d2b1bf2a3f8aa58322","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-29 16:25:57.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"0d8d01e7db6c18cb043f63aa761c24a7c047c784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 18:08:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64f9bf9cb91240a79cad769996c224e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c807df973e4869846e2a968aa71c44 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/770a513550a6405aaa9ccdb94cec6222 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60a41748e47a4d38af6ab86fc20eeb08 : SUCCESS in 8m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dac4d2ef4e11452c91039e186f974094 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99fdb214dec48d69555515461233128 : SUCCESS in 11m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07b6690fdc7d489595edd4fbd8dcc61e : SUCCESS in 7m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9539f5423b9041c08c3a60768ba6518b : SUCCESS in 1h 25m 07s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ce29aaed5fb84ec3ab35a7984b1b67b9 : FAILURE in 10m 44s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6e1f63ceb4be4675ba0982c62a30f236 : SUCCESS in 1h 26m 13s","accounts_in_message":[],"_revision_number":7},{"id":"e135f841faf261313c65d1b183d4fa7c1b8ec4c0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-17 13:43:18.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"a275a0e50054edc39f9db8bd8bc66fa14fbfc803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:43:52.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3b4ced426efb0f5c879d38fe3327cb0e2ce9f597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:13:05.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4116e041a1ee48cf8593fb1132ffc6e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee496444f9c841ce85b3ef7234017cbc : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7fe4f6cd80c48539abb9f03a66837a8 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c91763784ae4a1fb2898850530e8493 : SUCCESS in 7m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a17d3160c8040cabb0434c91556c073 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/032c19aa1f154b20b002148f881a8b13 : SUCCESS in 7m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/00de02dc0adb4044844ec82123250a84 : SUCCESS in 1h 15m 01s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/418d36a66a8c47fc9c2e7f1343d3d035 : SUCCESS in 1h 23m 51s","accounts_in_message":[],"_revision_number":7},{"id":"195f7781879251312eb9ee493c1bf5d6126b8090","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:13:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"5df3c5187a85dbfd61d291dfaac21087ffc1eb16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:14:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5411005069147c48e43307d21890285\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1fb537176e49f5948f6a123ff082c9 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1e571c14ea6a9e03e6ed91af7594022457a2c2cb","revisions":{"2c7ab9f5ae45935e05e739e41adda2d479f1c548":{"kind":"REWORK","_number":1,"created":"2020-07-01 15:44:06.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/738872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/738872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/1"}}},"commit":{"parents":[{"commit":"34a05bb41183f7a11d1e5453e073c0406dd138ce","subject":"Implement \"modprobe\" using oslo.privsep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/34a05bb41183f7a11d1e5453e073c0406dd138ce"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-01 11:31:17.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-01 15:43:20.000000000","tz":120},"subject":"Implement \"brctl\" and \"bridge\" using oslo.privsep","message":"Implement \"brctl\" and \"bridge\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"brctl\" and \"bridge\"\ncommands and replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40260\nChange-Id: I21e1d5f1e446afffaba6002f277a2990f934de29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2c7ab9f5ae45935e05e739e41adda2d479f1c548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2c7ab9f5ae45935e05e739e41adda2d479f1c548"}]},"branch":"refs/heads/master"},"b8a89c99f71237ec85e2eed7b5e31b8a58c5cebd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-12 09:54:51.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/738872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/738872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/2"}}},"commit":{"parents":[{"commit":"95baacc856933d3c71bfc6b0cfcbcb3777ecc6dc","subject":"Implement \"modprobe\" using oslo.privsep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/95baacc856933d3c71bfc6b0cfcbcb3777ecc6dc"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-01 11:31:17.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-10-12 09:54:51.000000000","tz":0},"subject":"Implement \"brctl\" and \"bridge\" using oslo.privsep","message":"Implement \"brctl\" and \"bridge\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"brctl\" and \"bridge\"\ncommands and replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40260\nChange-Id: I21e1d5f1e446afffaba6002f277a2990f934de29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b8a89c99f71237ec85e2eed7b5e31b8a58c5cebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b8a89c99f71237ec85e2eed7b5e31b8a58c5cebd"}]},"branch":"refs/heads/master"},"b8af628bb1d7809b3ab9ecfbf0d62fb3115296f5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-10-13 09:14:57.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/738872/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/738872/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/3"}}},"commit":{"parents":[{"commit":"3c32b95b527ddaaaaebc1a8bb381b846283519a5","subject":"Implement \"modprobe\" using oslo.privsep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3c32b95b527ddaaaaebc1a8bb381b846283519a5"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-01 11:31:17.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-10-13 09:14:57.000000000","tz":0},"subject":"Implement \"brctl\" and \"bridge\" using oslo.privsep","message":"Implement \"brctl\" and \"bridge\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"brctl\" and \"bridge\"\ncommands and replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40260\nChange-Id: I21e1d5f1e446afffaba6002f277a2990f934de29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b8af628bb1d7809b3ab9ecfbf0d62fb3115296f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b8af628bb1d7809b3ab9ecfbf0d62fb3115296f5"}]},"branch":"refs/heads/master"},"f5d05bd17f10660148f122eafd8fe68f0b7e7a26":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-12-01 18:48:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/738872/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/738872/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/4"}}},"commit":{"parents":[{"commit":"fbda54e1601773f762439f7ce7c70ff97a5484b6","subject":"Implement \"modprobe\" using oslo.privsep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fbda54e1601773f762439f7ce7c70ff97a5484b6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-01 11:31:17.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-12-01 18:48:29.000000000","tz":0},"subject":"Implement \"brctl\" and \"bridge\" using oslo.privsep","message":"Implement \"brctl\" and \"bridge\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"brctl\" and \"bridge\"\ncommands and replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40260\nChange-Id: I21e1d5f1e446afffaba6002f277a2990f934de29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f5d05bd17f10660148f122eafd8fe68f0b7e7a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f5d05bd17f10660148f122eafd8fe68f0b7e7a26"}]},"branch":"refs/heads/master","description":"Rebase"},"3e0811c97d186c848aea532ed0e22d7dbd5e6c14":{"kind":"REWORK","_number":5,"created":"2021-03-10 07:34:07.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/738872/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/738872/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/5"}}},"commit":{"parents":[{"commit":"a7e479f043d31cbfe57f5b0039757fa8e3546547","subject":"Merge \"Set secure mode for mpls bridge\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a7e479f043d31cbfe57f5b0039757fa8e3546547"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-01 11:31:17.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-03-10 07:33:46.000000000","tz":60},"subject":"Implement \"brctl\" and \"bridge\" using oslo.privsep","message":"Implement \"brctl\" and \"bridge\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"brctl\" and \"bridge\"\ncommands and replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40260\nChange-Id: I21e1d5f1e446afffaba6002f277a2990f934de29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3e0811c97d186c848aea532ed0e22d7dbd5e6c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3e0811c97d186c848aea532ed0e22d7dbd5e6c14"}]},"branch":"refs/heads/master"},"88235621757157e38cf7a2bad654e6ecfdf2ef01":{"kind":"REWORK","_number":6,"created":"2021-03-10 07:39:13.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/738872/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/738872/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/6"}}},"commit":{"parents":[{"commit":"a7e479f043d31cbfe57f5b0039757fa8e3546547","subject":"Merge \"Set secure mode for mpls bridge\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a7e479f043d31cbfe57f5b0039757fa8e3546547"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-01 11:31:17.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-03-10 07:39:05.000000000","tz":0},"subject":"Implement \"brctl\" and \"bridge\" using oslo.privsep","message":"Implement \"brctl\" and \"bridge\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"brctl\" and \"bridge\"\ncommands and replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40260\nChange-Id: I21e1d5f1e446afffaba6002f277a2990f934de29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/88235621757157e38cf7a2bad654e6ecfdf2ef01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/88235621757157e38cf7a2bad654e6ecfdf2ef01"}]},"branch":"refs/heads/master"},"1e571c14ea6a9e03e6ed91af7594022457a2c2cb":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-03-29 16:25:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/72/738872/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/738872/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/738872/7"}}},"commit":{"parents":[{"commit":"3ce7bf753526eca7de141f0ceed250a4713a18d5","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3ce7bf753526eca7de141f0ceed250a4713a18d5"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-01 11:31:17.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-29 16:25:57.000000000","tz":0},"subject":"Implement \"brctl\" and \"bridge\" using oslo.privsep","message":"Implement \"brctl\" and \"bridge\" using oslo.privsep\n\nRemove the use of oslo.rootwrap when executing \"brctl\" and \"bridge\"\ncommands and replace them with oslo.privsep.\n\nStory: #2007686\nTask: #40260\nChange-Id: I21e1d5f1e446afffaba6002f277a2990f934de29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1e571c14ea6a9e03e6ed91af7594022457a2c2cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1e571c14ea6a9e03e6ed91af7594022457a2c2cb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
