)]}'
{"id":"openstack%2Fnetworking-bagpipe~739672","triplet_id":"openstack%2Fnetworking-bagpipe~master~I7df9fa0f8c1b4fe456c1ee8ed65d9a7d13cbe11a","project":"openstack/networking-bagpipe","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I7df9fa0f8c1b4fe456c1ee8ed65d9a7d13cbe11a","subject":"Change bagpipe jobs to focal","status":"MERGED","created":"2020-07-07 07:56:59.000000000","updated":"2020-09-16 21:55:58.000000000","submitted":"2020-09-16 21:42:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739672-1600292552024-b2ad87a5","meta_rev_id":"b9d053ba836a8ac7ecc58cc170850e0747bd487f","_number":739672,"virtual_id_number":739672,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 21:42:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2020-09-10 08:21:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2020-09-16 19:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2020-09-16 19:57:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 08:21:28.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-09-16 19:57:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-16 21:42:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a91762dc817ae210a245a9f4bafac1ecc143e91c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-07 07:56:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2063957d8b57ed1fed0e1c22c1e6dfeb16bf6e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 09:25:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3be2c7cf2714abca1f59e3ba5ff57cf : FAILURE in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e76327a14f5c4aba9ae1590434dacb12 : SUCCESS in 10m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0914be599e9840c9bb2c2850644bb58a : SUCCESS in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1025762e474bf39016b3a3dc46e2cc : SUCCESS in 11m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef00920be272436ca4fa0af7c336edd0 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/778d2b9177d2419284d79e6f0d0b2e9c : SUCCESS in 11m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9ff77d8bd8de4761a2cccbf63cb41317 : SUCCESS in 1h 26m 35s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2c0528d9b5154eaea183c2a44197c50a : FAILURE in 19m 16s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/58606608dbcc427590762849a2db239d : FAILURE in 1h 17m 42s","accounts_in_message":[],"_revision_number":1},{"id":"fb5f218eb4d9e39e4a1e1c4c2ae8dc24fbfa7c20","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-08 11:58:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"05a5750c32b3a302119c027b2d8bac82a7591e4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 13:44:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/814703aa5edc4583b65d0c0e95e1594c : FAILURE in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f79b5ef9a9534f20bfb0172cf662033e : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67f6b326602240f0a751f84e4350c5fb : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037a283fbc394e78af167018799d9eb3 : SUCCESS in 9m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9f79e5eb4d444e68981e578b8223645 : FAILURE in 9m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7614d0cc9a647a1a47ca3b784832bd9 : SUCCESS in 8m 36s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0dde0c1ba7d3414fb138292b5c93975f : SUCCESS in 1h 39m 49s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/6989bc8f940646a99eeef6e76f2a0586 : FAILURE in 22m 26s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4b7033b39de34be794ba34fd5ca0b74a : FAILURE in 1h 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"140e2e398a4fa9e3227afda8e097dc7824cb9cfb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-28 14:44:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"799e9a9798134a8b49d81936e3cdd32abd9d9de0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:33:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0b48af7fe884dfe89837b22b4c0d147 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfabcc10b6a34e478178f4b793d3c274 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/620715227baf4ea5a1201268026265a0 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e64eedd6614bedbbce2490dd51a7b6 : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9b271b51bda451a8763eb96ccd404cf : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5482db9329854ddca1c96f5bcb67c9f9 : SUCCESS in 9m 24s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b6148c10c0294f5ca1beca7f5a57a480 : SUCCESS in 1h 28m 00s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/3b70cd01bb804b1284d59ea634e344a4 : FAILURE in 21m 01s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/23fd33c7270c43d0b07b507306038e31 : FAILURE in 52m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5c5d4a846edd7a111b13ff7c830aabdf346cac9b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-29 09:19:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f265913c681d78b2bacbb9b97aad0901ef5bc666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 10:58:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e89b541fa55410d95ce908ec4e0d629 : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d2e669123b947459f3c65d35fba61d9 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f2e609c7d1d471680573d8378dd7e19 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/603462d29ce14aa7830bb9be18693228 : SUCCESS in 6m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bc0e4c7ad604da7bccebf654b12efa5 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/113b735d1b2e4ed3820f70dccf542967 : SUCCESS in 6m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/58c8961f8e2342e7a3354235544800fc : SUCCESS in 1h 20m 14s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/3fef2760fa7f48bdafbb864e81ff3f33 : FAILURE in 14m 56s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4541cfdbb53a4671914faadaf89e603b : FAILURE in 1h 35m 02s","accounts_in_message":[],"_revision_number":2},{"id":"29c0177554c026a5e7a9b3c3ff82336a8ec5ce83","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-04 14:36:22.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"81cc1cedcef7f99235380305455365237eac35cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 16:18:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ce285c864d417b89fdfa7a8861bc08 : SUCCESS in 10m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64fca89a661745a0b404bd5ce4b373b4 : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74309293c7714f8ea1805716e447848d : SUCCESS in 12m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6988074e7854138bcedea10cc2516ba : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c9a5cffcc2b49b1a02968d902761229 : FAILURE in 6m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fb6cc40bf344c92915d2ae2f99570ee : SUCCESS in 10m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/717e919da6e74e4e95a177ca2f369437 : SUCCESS in 1h 24m 03s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/172bee8e89bd4cc584dffc4939066042 : FAILURE in 23m 05s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3f224ae0d87b42c5abf87a03f2f0c25b : FAILURE in 1h 36m 46s","accounts_in_message":[],"_revision_number":3},{"id":"ca35a904738237f87a3c5ded141f10e9392bcb8b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-05 06:59:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"035db756c47656624398737182cd12b240324d52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 08:45:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/656e150c967649a49d0c5afa666a6e5b : SUCCESS in 10m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8707055c5cd84da29a8e3f7fdd202e2e : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bcd5c0eba8e446e8775760c3d18560d : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52caec63eb974afc87db47e387184300 : SUCCESS in 8m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/078c449f9e8e4074be2f315fb246de9b : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cead12176880406087d697ea97a05964 : SUCCESS in 10m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c3c5d7381f0b4999a3f7ab190159dafa : SUCCESS in 1h 29m 00s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f7fb3732aa4b4e5ba912639211ae5b1e : FAILURE in 23m 28s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0bce520919ff4b42893716fb5602de0d : FAILURE in 1h 13m 13s","accounts_in_message":[],"_revision_number":3},{"id":"77b49f97a3a4504c02ed2c9cc4b0b9fd370f2ab6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-27 13:50:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f6d9211800234c2395643c2b4eabfaa3d289e190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 15:31:42.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef3a3c6443d4a7dac5ff8e25a6affd7 : SUCCESS in 10m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3afd2a9f711471699e22ca24a98a828 : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66090b1eb02e45ac8135bc543d26f791 : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80ed84d83d634d1495ad46e8516d85c0 : SUCCESS in 8m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dce57b4af015431c98bad3fb5e82619e : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56eb963b54e3406b80f08ad69eacc85f : SUCCESS in 9m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ea177287d0184258bb64fe0cd877ab0e : SUCCESS in 1h 30m 53s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/3a94a9771ce742e093d6e99f3e54d604 : FAILURE in 16m 40s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4a5e824edd7042d4a646fdc2817a3c68 : FAILURE in 1h 12m 19s","accounts_in_message":[],"_revision_number":3},{"id":"f465ce19ee45395b0f8f3b141d4589c848a0a874","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-02 15:13:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3206bd1ba5f5868de0ac3065140efbc1c0f586ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 16:47:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750d2cd46ad2441b817dd63acc46a8a6 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2198fc44d417474aabe66bef6aad07c8 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab454f55a5f04372b9e53198ff64af3f : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e17fe49a81d74445bcc2ac7f48b3d32f : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c30e039221c4e29b72893e20605e91c : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80229dfc74584255963bee467e37beea : SUCCESS in 7m 41s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c8f6dae785dd4d92b486132b70b52e0c : SUCCESS in 1h 22m 56s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/dbd0e8f6b3284a64a6a2dcf24b413c6f : FAILURE in 15m 21s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/e7b8a7d675eb42d8af9c89343e7b15b2 : FAILURE in 1h 14m 30s","accounts_in_message":[],"_revision_number":3},{"id":"be83b394124ad6921aa4dd2416bbec1447e1b564","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-04 09:07:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad476fb6fbb932e6c649c9ff761e8e7a60819f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:29:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23225ba8bbb1433d8979000cae322aa3 : FAILURE in 9m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d0d860b4db44c87add7c54b1a70beda : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b61fe418d62f4be29e186cc81381a252 : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0268ad193c904ce596a0d3c070c21129 : SUCCESS in 11m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c80da6bdfbc34c5087ca2088ee5fc8ef : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9cdff3030454e8bb155761c4b3f962f : SUCCESS in 8m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/bfe3c4a8361f4cdaa72897a2a711877e : FAILURE in 25m 38s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2f88eaaf88bf4e82a3e0080df75fca2a : FAILURE in 14m 45s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4b8761e05347413eb34fd7d3e6aa6df9 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":4},{"id":"08aebdc1f350685a192fcf4a323febf26bcd5e38","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-07 08:00:07.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"39c2565ce17e115e048fa3e2e0e68bbc3a39a4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 09:31:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dbfbbaf797a4b5fa067f7ad25e67442 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/721aee4405934c50a46209aa2aac0643 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ebfc2436f194f8bb47255b0b585a531 : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a390285266445d981cdb5ae06f3e1e : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/417a17c61ab04f89be04ec5d60e62a9e : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e71ff1442a74ec2b08887795850df6d : SUCCESS in 7m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c109aaf791ed443899ef4eb2ece7e113 : SUCCESS in 1h 21m 47s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/7fb0c5270b4e491790b9d7263d189439 : FAILURE in 18m 53s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/222215e14a6a49d6a879cee341344292 : SUCCESS in 1h 16m 48s","accounts_in_message":[],"_revision_number":4},{"id":"e84889e03f6bc87dca253a0714feb0f376508096","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-07 13:22:07.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"3c69beb1edd509d25cb1174e0c95e383e6470635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 14:22:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09a3d8a912540beaa1c5ac63a7ff1eb : SUCCESS in 19m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9b90cf00be34596b68e782c3c0441eb : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b387ac95a4224ca0a8231075ea4f7e42 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5436bb8f74ef4bb4ab4b1581de1d4ddc : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ea1959d7e6949c4b5df9b7ae3f1b08c : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/657fcd17ca824722bf270a96f341ee43 : SUCCESS in 6m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9e5e239c25964d92ba36726635abdf35 : SUCCESS in 58m 07s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/b440b2cc7c5d4bd1814bea8e5b678a73 : FAILURE in 17m 27s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/fc53dde41f854459a6bd7beb0945ff12 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":5},{"id":"8f3432e5762b59196f423d5309042b36f793941f","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-09-10 08:21:28.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e05d508d4bca562ce17340192912062eb5c7c1e1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-16 19:57:45.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f548809bfc54a08c7587b73061a536512b201da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 19:58:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e866033b3fd3714db8bb9e2beb0b2c7a8ce45af4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:42:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bbcbe77a4b04729a1ea8b5aa3524b43 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eaa4048f0b84394ac0ad67a1050b434 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81450ab0d6044b3cb9689145d182e7fe : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80843dfe7f0f47cc91d38d37d976d70f : SUCCESS in 14m 26s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/69cef84d354344ee8abaed7ffa17bfb9 : SUCCESS in 1h 32m 36s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/5bc6d4bde337469fb69c27944ef3535c : SUCCESS in 1h 31m 47s","accounts_in_message":[],"_revision_number":5},{"id":"e01d089b9658e5be6022ed428e340aa7cf828d18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:42:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"05a8239d5daeaa7d452971a887c7ca42924d2127","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 21:55:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c47ab85e040540f7bf33bc5c141caedc : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2f30efb0614c1f3e289d39969854eb3ed872f4e8","revisions":{"7df1eb53731518a66bb1e4fad35bc022f3a69944":{"kind":"REWORK","_number":1,"created":"2020-07-07 07:56:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/739672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/739672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/1"}}},"commit":{"parents":[{"commit":"d27919f356cd9d00924245eb44159609a259a9ee","subject":"Merge \"Switch to hacking 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d27919f356cd9d00924245eb44159609a259a9ee"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-07 07:55:51.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-07 07:55:51.000000000","tz":120},"subject":"DNM: test bagpipe with focal","message":"DNM: test bagpipe with focal\n\nChange-Id: I7df9fa0f8c1b4fe456c1ee8ed65d9a7d13cbe11a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7df1eb53731518a66bb1e4fad35bc022f3a69944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7df1eb53731518a66bb1e4fad35bc022f3a69944"}]},"branch":"refs/heads/master"},"83080e37c87d51ec1404682f20ad91d86d46bba1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-28 14:44:47.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/739672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/739672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/2"}}},"commit":{"parents":[{"commit":"e5ffadc18cdb1dca579b08413ff072d3cebfdef4","subject":"Switch from unittest2 compat methods to Python 3.x methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e5ffadc18cdb1dca579b08413ff072d3cebfdef4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-07 07:55:51.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-07-28 14:44:47.000000000","tz":0},"subject":"DNM: test bagpipe with focal","message":"DNM: test bagpipe with focal\n\nChange-Id: I7df9fa0f8c1b4fe456c1ee8ed65d9a7d13cbe11a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/83080e37c87d51ec1404682f20ad91d86d46bba1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/83080e37c87d51ec1404682f20ad91d86d46bba1"}]},"branch":"refs/heads/master"},"38d28ae348c0a8aad96425c25de760c6e82d857e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-04 14:36:22.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/739672/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/739672/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/3"}}},"commit":{"parents":[{"commit":"e5ffadc18cdb1dca579b08413ff072d3cebfdef4","subject":"Switch from unittest2 compat methods to Python 3.x methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e5ffadc18cdb1dca579b08413ff072d3cebfdef4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-07 07:55:51.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-08-04 14:36:19.000000000","tz":0},"subject":"Change bagpipe jobs to focal","message":"Change bagpipe jobs to focal\n\nChange-Id: I7df9fa0f8c1b4fe456c1ee8ed65d9a7d13cbe11a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/38d28ae348c0a8aad96425c25de760c6e82d857e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/38d28ae348c0a8aad96425c25de760c6e82d857e"}]},"branch":"refs/heads/master"},"8178550037bd6409e1cfaf6af0a4aeaa57271c7f":{"kind":"REWORK","_number":4,"created":"2020-09-04 09:07:30.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/739672/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/739672/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/4"}}},"commit":{"parents":[{"commit":"e5ffadc18cdb1dca579b08413ff072d3cebfdef4","subject":"Switch from unittest2 compat methods to Python 3.x methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e5ffadc18cdb1dca579b08413ff072d3cebfdef4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-07 07:55:51.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-09-04 09:07:06.000000000","tz":120},"subject":"Change bagpipe jobs to focal","message":"Change bagpipe jobs to focal\n\nChange-Id: I7df9fa0f8c1b4fe456c1ee8ed65d9a7d13cbe11a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8178550037bd6409e1cfaf6af0a4aeaa57271c7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8178550037bd6409e1cfaf6af0a4aeaa57271c7f"}]},"branch":"refs/heads/master"},"2f30efb0614c1f3e289d39969854eb3ed872f4e8":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-09-07 13:22:07.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/739672/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/739672/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/739672/5"}}},"commit":{"parents":[{"commit":"e5ffadc18cdb1dca579b08413ff072d3cebfdef4","subject":"Switch from unittest2 compat methods to Python 3.x methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e5ffadc18cdb1dca579b08413ff072d3cebfdef4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-07 07:55:51.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-09-07 13:22:02.000000000","tz":0},"subject":"Change bagpipe jobs to focal","message":"Change bagpipe jobs to focal\n\nAdd some volume tests to tempest_black_regex till\nhttps://bugs.launchpad.net/nova/+bug/1882521 is solved.\n\nChange-Id: I7df9fa0f8c1b4fe456c1ee8ed65d9a7d13cbe11a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2f30efb0614c1f3e289d39969854eb3ed872f4e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2f30efb0614c1f3e289d39969854eb3ed872f4e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
