)]}'
{"id":"openstack%2Fnetworking-bagpipe~740155","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ia86c627602deab43b695f60568927e737acf3417","project":"openstack/networking-bagpipe","branch":"master","hashtags":[],"change_id":"Ia86c627602deab43b695f60568927e737acf3417","subject":"Switch from unittest2 compat methods to Python 3.x methods","status":"MERGED","created":"2020-07-09 04:59:11.000000000","updated":"2020-07-20 18:05:10.000000000","submitted":"2020-07-20 18:03:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740155-1595268210705-9ed56beb","meta_rev_id":"0190e0326a9e46a251eaa800e2e29bdb01344443","_number":740155,"virtual_id_number":740155,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2020-07-13 08:15:25.000000000","_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 18:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2020-07-10 13:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2020-07-10 13:40:41.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2020-07-10 13:42:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-10 13:40:41.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-07-10 13:42:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-07-20 18:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad379da87b722c0c3908eea1357aee840f71a369","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-07-09 04:59:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b13f2f35d5a8270c5c1364c4fb79028d35eec5c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 06:24:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc3a5ee84f24c0bbf7b5c54af9bcf4f : FAILURE in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9ccdc9ea28541e898a05a77d499140f : SUCCESS in 14m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa02786f1c834b3a9da17737a9a71d24 : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8037b6e50c6d41bba8acc69d351fa33a : SUCCESS in 11m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/870465dd8ec34e1dbf47099f8a5e7986 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cf6b3d8b0e44e919154d01cae763cba : SUCCESS in 8m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9bedf47d54c2434b958add4d82282f00 : SUCCESS in 1h 24m 33s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/fae65d0de45e446cb6eac295d0bf7e74 : FAILURE in 18m 00s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/d7df0ae22ee64d0fa49aade34e9a1e4c : SUCCESS in 1h 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9467ec61b854fb665bfb75384ada9c52c7c3c86f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-10 11:45:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5215fdc19a971dfa62204e5eb984df7b8155414a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 13:15:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13e77c71a58b41eaaf0b7594b6a0d0b2 : SUCCESS in 9m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa0facc46cab41eab6328ea42757592d : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c16de0913be44401b50a66d2e7dc0570 : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac43300d35d4bf5807cb44e61bd9a68 : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8d44596e3d7443590cb0c5d453b9c25 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/816fb0315c994a0db3a47faf1900dc85 : SUCCESS in 11m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/482ff8844f9c4ae59df73d0480080004 : SUCCESS in 1h 27m 35s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/37f7c66bbb2e406e8480a8e07ff7cc9b : FAILURE in 20m 37s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b137ee7727a54764a8e591605237db76 : SUCCESS in 1h 14m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f6f59e15ff9f1dbf86425da556102ada3c7a21e6","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-07-10 13:40:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"04f332afa040e51bfb399809060967a01dd1676d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-10 13:42:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nthanks for taking care of it","accounts_in_message":[],"_revision_number":2},{"id":"66e2625c77a3d11fdc8dd4b755f0b99a708ad78c","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-07-13 08:15:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8ed226c3ab225d1e76cfb58cc95f8dc44df1277e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 09:49:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5362402e1fe347c7a8a455f3ca5a2df3 : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da884d51d202465d9cdb20d893fdba46 : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80b238ef5753477bb7096044ff2fce30 : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180e344c081f472791df196d12dd9540 : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b18a7772b02c47c39147b95141ba13d2 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5432579b4d9f41708009a0c9624a30b6 : SUCCESS in 11m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4c25da7699f44552a878a84bdc32b98f : SUCCESS in 1h 33m 06s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/8eb8e61fcfda4742ae26a66fa02a048d : FAILURE in 19m 34s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ed2c0c1cc4a2489daa356e977f50e087 : FAILURE in 1h 09m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e7ad2a8db271b528c18c1db9d100f74de5b67af3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-13 11:51:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3ea117e03db15f96d4d325642a8fe4d1983b79ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 13:29:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ec0f1cf33f4f158dedb116e50a7690 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ecc86dda96a4f778d45836558ee644b : SUCCESS in 10m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e861776f7cb4a3ea97897f8f5442c99 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a335076f75464f6595ba7ffe552405a9 : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dddcc3c138541a4995989b5a8eab1f1 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fdbdb6ddbdf4ed8b44f00921b9564c4 : SUCCESS in 10m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2c958253b5784fb086f44bfbfb976ede : SUCCESS in 1h 35m 52s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/5ce861bad81145f6a67c439e74c1f694 : FAILURE in 19m 38s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c84a9426828545d3a814f29f530eff0f : FAILURE in 1h 15m 41s","accounts_in_message":[],"_revision_number":2},{"id":"2dfaaf7c4e997c4ca89af716ca77a4764289540d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-14 08:50:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d2bbc72fe60e549474a8bd700c49c16a1aa34705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 10:28:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc0d843686546b7b2b5ab03642286c3 : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/561647017bf94566bb10675bfc9153f3 : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fbe97886d934ccc9d6a8e1ec05868f9 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b50eeb7c3c46ba9b51330cf34ea879 : SUCCESS in 11m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27985aa8eeb140f39d71bccdb06e3fd5 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9c8828fe1354c12919b4e44936d33d5 : SUCCESS in 9m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8d59f148a1494f44bede04f54cd18d3f : SUCCESS in 1h 30m 05s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/7f690ad8849549748d103dc0d4064c70 : FAILURE in 15m 15s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/64f3bede8975467fb2cd4588d9a9dafa : SUCCESS in 1h 26m 35s","accounts_in_message":[],"_revision_number":2},{"id":"249fe87cf5112fded1aff01e344546ab888d5046","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 16:39:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3cc6a17b9e2c505bc633de6a82433654c3eed37c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:03:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4bf18eafae4ba69c06deaa46770a5d : SUCCESS in 9m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e531440840241ab871c83140d81365d : SUCCESS in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecc5233589a0469aa14b55f291eaff85 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ad8d9cb5444ce5b9d5e7c9f5b6235d : SUCCESS in 8m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/981cf3b43e1842b19e4a94c34b556e3b : SUCCESS in 1h 10m 35s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/adedb919985141b0b1d0229b2a942b1a : SUCCESS in 1h 17m 26s","accounts_in_message":[],"_revision_number":2},{"id":"443a60410934149e5d2193a4d45361ccc3287cb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:03:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0190e0326a9e46a251eaa800e2e29bdb01344443","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:05:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345a85bfff534ce1b4a1d73bbf87c9d9 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5ffadc18cdb1dca579b08413ff072d3cebfdef4","revisions":{"1e5f626847b3e0eda2b5567c6804500a69cee09b":{"kind":"REWORK","_number":1,"created":"2020-07-09 04:59:11.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/55/740155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/55/740155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/740155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/740155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/740155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/740155/1"}}},"commit":{"parents":[{"commit":"d27919f356cd9d00924245eb44159609a259a9ee","subject":"Merge \"Switch to hacking 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d27919f356cd9d00924245eb44159609a259a9ee"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-07-09 04:57:17.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-07-09 04:57:17.000000000","tz":480},"subject":"Switch from unittest2 compat methods to Python 3.x methods","message":"Switch from unittest2 compat methods to Python 3.x methods\n\nWith the removal of Python 2.x we can remove the unittest2 compat\nwrappers and switch to assertCountEqual instead of assertItemsEqual\n\nWe have been able to use them since then, because\ntesttools required unittest2, which still included it. With testtools\nremoving Python 2.7 support [3][4], we will lose support for\nassertItemsEqual, so we should switch to use assertCountEqual.\n\n[1] - https://bugs.python.org/issue17866\n[2] - https://hg.python.org/cpython/rev/d9921cb6e3cd\n[3] - testing-cabal/testtools#286\n[4] - testing-cabal/testtools#277\n\nChange-Id: Ia86c627602deab43b695f60568927e737acf3417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1e5f626847b3e0eda2b5567c6804500a69cee09b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1e5f626847b3e0eda2b5567c6804500a69cee09b"}]},"branch":"refs/heads/master"},"e5ffadc18cdb1dca579b08413ff072d3cebfdef4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-10 11:45:51.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/55/740155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/55/740155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/740155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/740155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/740155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/740155/2"}}},"commit":{"parents":[{"commit":"30b058ec04f4ccbbbca41c36b72f877634873352","subject":"Fix pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/30b058ec04f4ccbbbca41c36b72f877634873352"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-07-09 04:57:17.000000000","tz":480},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-07-10 11:45:51.000000000","tz":0},"subject":"Switch from unittest2 compat methods to Python 3.x methods","message":"Switch from unittest2 compat methods to Python 3.x methods\n\nWith the removal of Python 2.x we can remove the unittest2 compat\nwrappers and switch to assertCountEqual instead of assertItemsEqual\n\nWe have been able to use them since then, because\ntesttools required unittest2, which still included it. With testtools\nremoving Python 2.7 support [3][4], we will lose support for\nassertItemsEqual, so we should switch to use assertCountEqual.\n\n[1] - https://bugs.python.org/issue17866\n[2] - https://hg.python.org/cpython/rev/d9921cb6e3cd\n[3] - testing-cabal/testtools#286\n[4] - testing-cabal/testtools#277\n\nChange-Id: Ia86c627602deab43b695f60568927e737acf3417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e5ffadc18cdb1dca579b08413ff072d3cebfdef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e5ffadc18cdb1dca579b08413ff072d3cebfdef4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
