)]}'
{"id":"openstack%2Fnetworking-bagpipe~740286","triplet_id":"openstack%2Fnetworking-bagpipe~master~I0cc70af50dbbad972baece017c7b31e0ae696579","project":"openstack/networking-bagpipe","branch":"master","topic":"pep8-error","hashtags":[],"change_id":"I0cc70af50dbbad972baece017c7b31e0ae696579","subject":"Fix pep8 job","status":"MERGED","created":"2020-07-09 13:26:34.000000000","updated":"2020-07-20 17:59:40.000000000","submitted":"2020-07-20 17:56:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740286-1595267797903-b2062de1","meta_rev_id":"a39f59e8187e4a25b23912b2d5dd6851435fe285","_number":740286,"virtual_id_number":740286,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 17:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2020-07-18 10:46:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2020-07-15 10:25:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-07-20 16:39:00.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-07-20 16:39:00.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-07-10 11:32:42.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-10 11:32:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-07-15 10:25:24.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-18 10:46:01.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-07-20 16:39:00.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-07-20 17:56:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60ead21eae5c571188cb958faed12ff2aeca9e73","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-09 13:26:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"126363b96607d379a7798c268bf049ba2b7903c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 15:09:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c12cdabea4c4524b6cd924ed6cc39cb : SUCCESS in 11m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc5e9917df664ff0b1d35c7203830998 : SUCCESS in 12m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d0d58169b4046298d40bf4f53b66dfc : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfa3bf46c67641d9a2877bf97a59469c : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/718d5d8774c84dab888660389092412d : SUCCESS in 4m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4c19e783b3e54220bea07742f78462a3 : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4742ecd8dcb84b0281259fa1276b61d1 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd83b52aae6647d4b5d2e59bdfc1bd23 : SUCCESS in 11m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/72336f919088443bafaf17204ca278ca : SUCCESS in 1h 14m 42s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/c45566ed308949de83a34916e9090128 : FAILURE in 23m 50s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b05f9acf5a8c4624a2e21fbc4171b65a : SUCCESS in 1h 37m 39s","accounts_in_message":[],"_revision_number":1},{"id":"798d1a62c00aa4223364b2d24b0dbd74bbc1ed4b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-15 10:25:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"63886409bd0a32c5a17714505217890a8ee25f26","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-07-18 10:46:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e125a6da9a44dbadd10625de24858d8faa059c28","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-07-20 16:39:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c520c13febf1367088ace384e6ae4b95ff896691","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 16:39:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eaea0f4218fb8ae7a0ec763c2ebee7e564b329e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 17:56:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/802af83d44c54438ad53b2ce5941fb88 : SUCCESS in 10m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08cf989b2faa4a31ae3e004a78fad1e0 : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/952cefb3db564f6399648542f896c9c1 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4357a77cfa0f4ba68e64625bec8f9b67 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adcaf5e82ade485d982a91f6ca999710 : SUCCESS in 5m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bbcdc566f84f4125a94c5c37677e7279 : SUCCESS in 4m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c8a969e8a6ad4bf982f0a0529375a3f7 : SUCCESS in 1h 12m 54s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0b51166b95c94eeaaae51de559a5453a : SUCCESS in 1h 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f7d2c0ebdadf60b22f8ecaaef7e4f410d8ce3167","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 17:56:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cc35b640f60fdb845d4b453c16aa38cf5939c201","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 17:59:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adcbafa4150047b48fbe11becb424efc : SUCCESS in 2m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6b77ce993ce44779c680bdf35a3ec72 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30b058ec04f4ccbbbca41c36b72f877634873352","revisions":{"30b058ec04f4ccbbbca41c36b72f877634873352":{"kind":"REWORK","_number":1,"created":"2020-07-09 13:26:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/86/740286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/86/740286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/86/740286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/86/740286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/86/740286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/86/740286/1"}}},"commit":{"parents":[{"commit":"d27919f356cd9d00924245eb44159609a259a9ee","subject":"Merge \"Switch to hacking 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d27919f356cd9d00924245eb44159609a259a9ee"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-09 13:25:55.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-09 13:25:55.000000000","tz":120},"subject":"Fix pep8 job","message":"Fix pep8 job\n\nNew versions of isort broke pylint. This patch fixes it at 4.3.21.\n\nChange-Id: I0cc70af50dbbad972baece017c7b31e0ae696579\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/30b058ec04f4ccbbbca41c36b72f877634873352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/30b058ec04f4ccbbbca41c36b72f877634873352"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
