)]}'
{"id":"openstack%2Fnetworking-bagpipe~743211","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fussuri~I4d21cde54a475ae0d259f292dccdda51e9760d0b","project":"openstack/networking-bagpipe","branch":"stable/ussuri","topic":"DNM","hashtags":[],"change_id":"I4d21cde54a475ae0d259f292dccdda51e9760d0b","subject":"DNM: test stable/ussuri branch","status":"ABANDONED","created":"2020-07-27 12:23:19.000000000","updated":"2020-08-24 16:10:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ead2af4f07f25b4df642865e49666d883b542ee2","_number":743211,"virtual_id_number":743211,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-27 16:10:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 16:10:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a08d824eb424dc99fc2fc27fa4ea36f36273bb01","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-27 12:23:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e29088f64e215d6aa703aff2ec5f36272fda718e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 14:11:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ef126853714968bbd95f303ca1e170 : FAILURE in 10m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/392337970e2848e897ca321428e40c95 : SUCCESS in 11m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b48da573b896402dbe31761eae95ddd3 : SUCCESS in 10m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56ecbc585e5e4df6b13ef5eee5db4c78 : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e08e488c834f46b614e41781d7a1f2 : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68a7c2cdcc5e421e803e95c06fd5a381 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fcf8f76ed554aff967243ec6ea10d1a : SUCCESS in 11m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/d3617ced83f645b9ad1479405265f921 : SUCCESS in 1h 33m 34s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ebde0bb103704d6f82aeb360b9ea1868 : FAILURE in 22m 01s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0120d0f11c6d484295dd1c42bc3e3211 : SUCCESS in 1h 24m 02s","accounts_in_message":[],"_revision_number":1},{"id":"728c9aaa0845ec3a566b12afb03b14699560fd3f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-27 14:41:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"68d97d06b779c10a973980091c12feb1b9c9d406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 16:10:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b6ae6253434f2db24d77ce76ef5ba1 : SUCCESS in 9m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e39863b41d2245cc9c8c79de16361d04 : SUCCESS in 10m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ead9a0f421d4eedbae020863e6b653f : SUCCESS in 10m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5f82d42112149f99e0726a47c338c8d : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec4cc604ad548c6b81b0bc6fc946115 : SUCCESS in 10m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c46468a4a6749fdbf24b81133f50050 : FAILURE in 11m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20a29c0e46c841d29a422090b7e22b30 : SUCCESS in 11m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-ussuri https://zuul.opendev.org/t/openstack/build/833733be40414d9390593ba7c0812406 : SUCCESS in 1h 17m 45s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/13db782768a94b739341b1ebb7c03d50 : FAILURE in 15m 34s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/337cb67155d440038d36b8abbfbba32a : SUCCESS in 1h 24m 15s","accounts_in_message":[],"_revision_number":2},{"id":"46cfbf928e9f6f742f96f7789e1b59aee156e02d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-24 16:10:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a566a6be3ec0be701552c37ce9eadff0cc94215b","revisions":{"e867f8c9d74f003c9976e33d386be963be686aaa":{"kind":"REWORK","_number":1,"created":"2020-07-27 12:23:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/11/743211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/11/743211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/11/743211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/11/743211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/11/743211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/11/743211/1"}}},"commit":{"parents":[{"commit":"d6bee053b2523f94ec621f83237f02acafb42e27","subject":"Switch to stable/ussuri neutron-tempest-plugin jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d6bee053b2523f94ec621f83237f02acafb42e27"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-27 12:22:29.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-27 12:22:29.000000000","tz":120},"subject":"DNM: test stable/ussuri branch","message":"DNM: test stable/ussuri branch\n\nChange-Id: I4d21cde54a475ae0d259f292dccdda51e9760d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e867f8c9d74f003c9976e33d386be963be686aaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e867f8c9d74f003c9976e33d386be963be686aaa"}]},"branch":"refs/heads/stable/ussuri"},"a566a6be3ec0be701552c37ce9eadff0cc94215b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-27 14:41:37.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/11/743211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/11/743211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/11/743211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/11/743211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/11/743211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/11/743211/2"}}},"commit":{"parents":[{"commit":"b14e767b1db4ccd9a61d06fe74ce6cbd501fc3d6","subject":"Fix pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b14e767b1db4ccd9a61d06fe74ce6cbd501fc3d6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-27 12:22:29.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-07-27 14:41:37.000000000","tz":0},"subject":"DNM: test stable/ussuri branch","message":"DNM: test stable/ussuri branch\n\nChange-Id: I4d21cde54a475ae0d259f292dccdda51e9760d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a566a6be3ec0be701552c37ce9eadff0cc94215b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a566a6be3ec0be701552c37ce9eadff0cc94215b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
