)]}'
{"id":"openstack%2Fnetworking-bagpipe~747117","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Ftrain~I0cc70af50dbbad972baece017c7b31e0ae696579","project":"openstack/networking-bagpipe","branch":"stable/train","topic":"pep8-error-stable/train","hashtags":[],"change_id":"I0cc70af50dbbad972baece017c7b31e0ae696579","subject":"Fix pep8 job","status":"ABANDONED","created":"2020-08-20 07:51:44.000000000","updated":"2021-01-14 09:16:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"642b6e03725d4be71856c4251696ba54fa490919","_number":747117,"virtual_id_number":747117,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-20 09:54:08.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-20 09:10:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-08-24 12:03:42.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 09:10:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 12:03:42.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"023afc5f53a41e82195e95aac2847d5d57448c78","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-20 07:51:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"ece2179e5d87f49fe083635c763c754eac70e415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 09:10:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a3fd0cd8e1a4b779ff646f96c782038 : SUCCESS in 7m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97dc79ec6de14e72a13b2276607bc6bc : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57212bd9afb8430daeecb5be12e07c0d : FAILURE in 8m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e35e2a882ca84aa8bc857c8107fba4c2 : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e604072f98041a89951186c3687566c : FAILURE in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18086b2de14e4e819a6cd021bd93f7ed : SUCCESS in 6m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b77cd6cc6691454086d94bcc423312a0 : FAILURE in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01ca19b0f0f24f8d8f79ef3d8ce8a53c : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01c15d5f7f414c7e88f07f832dbdfdca : SUCCESS in 9m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/70fba20eb8de4687a8bd4f029c1e5ecd : SUCCESS in 1h 15m 35s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/23d82f0aca064a26a8c5011e6e43a8a2 : FAILURE in 10m 40s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1f37d36da8084f0885aec409d9eaf01a : SUCCESS in 1h 16m 49s","accounts_in_message":[],"_revision_number":1},{"id":"902dddd7877b99716a24f8627cfc0f98f0a19085","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-20 09:54:08.000000000","message":"Patch Set 1:\n\nLooks like train needs a bunch of fixes :(","accounts_in_message":[],"_revision_number":1},{"id":"041952d48f133911f2f1ec1ee22706382f120e8b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-24 12:03:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f18af43793490ae7267c6f39462de55a8edee286","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-24 12:03:42.000000000","message":"Patch Set 1:\n\n\u003e Looks like train needs a bunch of fixes :(\n\nI have seen similar unit test failures in stable periodic jobs","accounts_in_message":[],"_revision_number":1},{"id":"b5b0d48a2f916e6399b92c11e6b1d36ebdfffa78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 13:49:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fa31f084db641df8c18e935f18f6f94 : SUCCESS in 7m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f22028af6ae94a9cb37e8c7c2078b20e : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7146d5072ae42d28f0183663888b88e : SUCCESS in 7m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b5a2242bcac4439b8e920448805eaba : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6878a918fb5440392a680918d69495d : FAILURE in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d76ccf55f18b47acbc2162c1c47d2e68 : SUCCESS in 5m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e9dd36a15584c759df247aa633bfd25 : FAILURE in 3m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cff7abdcfee04ceebd55d3684fde54f0 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb384c2541254736b6cf085491a98c76 : SUCCESS in 9m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/0aaca982bd7243e1943fb9e43e1b450b : SUCCESS in 1h 27m 58s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/adba764d24724db6944b2df69acb3da1 : FAILURE in 13m 55s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/701c604b71074d5ca0c2f92ecc94e8ef : SUCCESS in 1h 29m 41s","accounts_in_message":[],"_revision_number":1},{"id":"642b6e03725d4be71856c4251696ba54fa490919","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 09:16:54.000000000","message":"Abandoned\n\nFix merged in https://review.opendev.org/c/openstack/networking-bagpipe/+/770450","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f903bff2eed5a80eb2472b7469a610e8f8aa00eb","revisions":{"f903bff2eed5a80eb2472b7469a610e8f8aa00eb":{"kind":"REWORK","_number":1,"created":"2020-08-20 07:51:44.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/17/747117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/747117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/747117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/747117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/747117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/747117/1"}}},"commit":{"parents":[{"commit":"613ba308588fbb3145e6773146fef65829bfae29","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/613ba308588fbb3145e6773146fef65829bfae29"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-09 13:25:55.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-08-20 07:51:44.000000000","tz":0},"subject":"Fix pep8 job","message":"Fix pep8 job\n\nNew versions of isort broke pylint. This patch fixes it at 4.3.21.\nConflicts: test-requirements.txt\n\nChange-Id: I0cc70af50dbbad972baece017c7b31e0ae696579\n(cherry picked from commit 30b058ec04f4ccbbbca41c36b72f877634873352)\n(cherry picked from commit b14e767b1db4ccd9a61d06fe74ce6cbd501fc3d6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f903bff2eed5a80eb2472b7469a610e8f8aa00eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f903bff2eed5a80eb2472b7469a610e8f8aa00eb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
