)]}'
{"id":"openstack%2Fnetworking-bagpipe~753632","triplet_id":"openstack%2Fnetworking-bagpipe~master~I8926a9cb9bec4151bc974e830c2460b336a09ba0","project":"openstack/networking-bagpipe","branch":"master","topic":"reno-victoria","hashtags":[],"change_id":"I8926a9cb9bec4151bc974e830c2460b336a09ba0","subject":"Update master for stable/victoria","status":"MERGED","created":"2020-09-23 12:46:03.000000000","updated":"2020-10-08 23:17:36.000000000","submitted":"2020-10-08 23:09:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753632-1602198591657-14270ba4","meta_rev_id":"966a630f2ca65576b5ee76d9d3d2b9b26aa56d12","_number":753632,"virtual_id_number":753632,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-08 23:09:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2020-10-08 07:42:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2020-09-25 12:38:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-10-08 19:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-09-25 12:11:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":1,"date":"2020-10-08 07:42:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 12:11:20.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-09-25 12:38:39.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-10-08 07:42:05.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-10-08 19:58:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-08 23:09:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"395ffc4543abdb15f97e4c05a9c97b66af0b8ac4","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-23 12:46:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5fe6ae41c2bc4d3c21e138116a14a9b39502f8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 14:31:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a0eebb3afe4c32a86376c529bc4c19 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e4eb5ae4aae4b03a29b66e749aac492 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65901f797bcf4bdcac41ef417646b136 : SUCCESS in 4m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b9af40cb92624fe39c22e2c6b6914697 : SUCCESS in 1h 13m 28s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/672c5835cfbe47b7bdf815a9b7369f97 : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ae4b5ea2a691caefbb7047712bdf75e3efea7c57","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-09-25 12:11:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"331548c6b74b6b44e56f557c18f55fe102832ad3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-25 12:38:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9abf77d2f945a2d2316ef1a0540c440a837f90e0","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-10-08 07:42:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbbdf15adce4f5715e0488399b8a38151796bcf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 07:42:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01518be7190373e290b626e0fc746f69f43cc949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 09:18:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edb18deb16ad4775a2f4a8855101d445 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9170641bd0d84e49b9796a84a1b5073a : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a2cd1a1664744919e9e1cc85fea2fb1 : SUCCESS in 4m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4f5b8ab3d34044a4a800bf357580b387 : SUCCESS in 1h 29m 46s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/53148df51d7a443e9e54f9c4b42c475e : FAILURE in 1h 08m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c68b459353164ead2bb7dbce760f2dd8b57ffe2f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-08 10:12:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b091b0750d8a1c89dd05c3e409d912d0eb295969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 11:44:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/996f1622567342208f02122bf8f8e450 : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71fa12088adc4409a0e6ba1635e4b127 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21792f90d3fd4a5fa79cc9cde14ee6e0 : SUCCESS in 4m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b141de03364e4f48bda8e64d995e2d95 : FAILURE in 1h 26m 31s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6e2641a2ecc0434089fc9d7144a3e6de : SUCCESS in 1h 20m 30s","accounts_in_message":[],"_revision_number":1},{"id":"415f43951e2f392388ffd72851e159f17266a052","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-10-08 13:28:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b254dee2918b9ff6d10e492da6dc43d5d452a715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 14:57:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28dc20815d094cdcb39ba385d55e1260 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb8f3b2fa044531bc74b89c3a4668f5 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/714506b8dbd64e57b1fb9169a7a97cfc : SUCCESS in 4m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e8f945dc6d644dfd8244756a50784b3e : FAILURE in 1h 24m 24s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/91b8be2363db4f33952b633136ff3260 : SUCCESS in 1h 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0a4e0e28fe036bb23c58bba52a209285e32316d9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-08 19:58:53.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d040b360761f05310a98779bd26b31f75483610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 21:30:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c378c67abb41c79467c721ad16df8f : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0b17a11aaf454cb9cd0ecd94bea959 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8ae1f5b8f5449fd8ddc5b80b576b3ac : SUCCESS in 3m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7bd25a0d2fde4069a75c52dc82cfe4e4 : SUCCESS in 1h 22m 54s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4c990a0b3fd14774aae720f0b3e8e097 : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7771e86b909b89d6683fdab4a240f9c376762ea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 21:31:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7eb9e17e50c9ee7e703c1f4c6fda6db25ff29336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 23:09:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/758b2841ae764e108eded97983167dcb : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042862393b0443548f8ee819a32c1117 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73181fbcd08b49b7bc128a792e43d18d : SUCCESS in 4m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/466db6c6093242acb240180eba2ed8c0 : SUCCESS in 1h 29m 00s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/314b581964154ea28eb9808f9688d009 : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"03d0549348d4a1901b839ba85829aa6833a3cd15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 23:09:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"51322a10498a1fa9177db0ef4c560b1f4b26a1be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 23:17:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f82cdb16c9f3487e9398d4a66d2fb0e8 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2babd0630dea4de2b0992065b92207f5 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bcd7446481e76fd5af3175c1b6cfbea99c57103","revisions":{"5bcd7446481e76fd5af3175c1b6cfbea99c57103":{"kind":"REWORK","_number":1,"created":"2020-09-23 12:46:03.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/32/753632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/32/753632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/32/753632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/32/753632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/32/753632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/32/753632/1"}}},"commit":{"parents":[{"commit":"2f30efb0614c1f3e289d39969854eb3ed872f4e8","subject":"Change bagpipe jobs to focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2f30efb0614c1f3e289d39969854eb3ed872f4e8"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-23 12:45:59.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-23 12:45:59.000000000","tz":0},"subject":"Update master for stable/victoria","message":"Update master for stable/victoria\n\nAdd file to the reno documentation build to show release notes for\nstable/victoria.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/victoria.\n\nChange-Id: I8926a9cb9bec4151bc974e830c2460b336a09ba0\nSem-Ver: feature\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5bcd7446481e76fd5af3175c1b6cfbea99c57103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5bcd7446481e76fd5af3175c1b6cfbea99c57103"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
