)]}'
{"id":"openstack%2Fnetworking-bagpipe~770450","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Ftrain~I321dcffc1ab2494de4bc398dfc97e5b5a6666495","project":"openstack/networking-bagpipe","branch":"stable/train","topic":"oslo_lc_drop","hashtags":[],"change_id":"I321dcffc1ab2494de4bc398dfc97e5b5a6666495","subject":"Dropping lower constraints testing (stable Train)","status":"MERGED","created":"2021-01-12 16:31:32.000000000","updated":"2021-01-21 13:56:58.000000000","submitted":"2021-01-21 13:43:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770450-oslo_lc_drop","meta_rev_id":"cbf84a3bac4fd94a4f782433ddc1524e456418f1","_number":770450,"virtual_id_number":770450,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 13:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-01-21 12:15:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-01-20 11:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-01-21 07:22:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-01-21 12:15:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 18:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-14 09:16:39.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-01-14 09:37:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-01-14 09:53:03.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-20 11:18:48.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-01-21 12:15:11.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"46a35506ed3ff006e069d3821c93580550ce7f50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 16:31:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5738cb836a4890f0cf01e05f10edc8406f17d1ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 18:38:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25efd4a5268d423ea35530ca6abae3f3 : FAILURE in 10m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f03e1b11dc1a431bb14a051e21dffb19 : SUCCESS in 10m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89cfa2d1aaeb4c21b9031d468434675c : SUCCESS in 10m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4e463b7964c4072a1892555b2689273 : SUCCESS in 16m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6244857d212a4e598f2ae08fc040f123 : FAILURE in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a2f8dd6cb154af9872103a22253a256 : SUCCESS in 8m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0cd312f3b1ad47e683fbeb74653320fd : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e89674e105b4567a885c9619aecbe5d : SUCCESS in 11m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/64809e3143034f90b5e1da58b0a86485 : SUCCESS in 1h 33m 08s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/02d75463ccdc472baea04044d51bebab : FAILURE in 16m 57s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/69132fc3686c4496936d6b2853f08d07 : SUCCESS in 1h 41m 35s","accounts_in_message":[],"_revision_number":1},{"id":"64879dd51bc0bbc873fd039fa092ff27aa2c2ddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 09:14:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d0ee6bc1a23cbf5e2a251eae976df801a85597a2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 09:16:39.000000000","message":"Patch Set 2:\n\nMerging the isort fix from https://review.opendev.org/c/openstack/networking-bagpipe/+/747117 it will still need a doc build fix though","accounts_in_message":[],"_revision_number":2},{"id":"4e8b3b0d72320e449ceca90e846b7156b5d072ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 09:53:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f4e93ef217ec751847c20e8bbbb3f15a64448358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:20:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8fdbe9ff0341d082406a96e19f02cf : SUCCESS in 8m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2697f4b392c04a868a8cd00c56234f6e : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e952c67e4c54413fa1f5b94eb39e5da4 : SUCCESS in 15m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0c08caf87bc4f93a620e783065973fa : SUCCESS in 19m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2caa0756e61240dabc15282d79061e00 : SUCCESS in 13m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd803677aad5487b947a0fa2e3d306e3 : SUCCESS in 10m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c022f588e34d49aa8e28ec59a3efe4b7 : FAILURE in 7m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/948c8c52aa82436080185310565af8db : SUCCESS in 16m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/d7ccddfdf1e948749855379578f50968 : SUCCESS in 1h 34m 40s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/257a993d818745178adeb8f30067879b : FAILURE in 18m 24s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/877e91d1bf1f4627892be171c6120fc8 : SUCCESS in 1h 43m 22s","accounts_in_message":[],"_revision_number":3},{"id":"add49331465868dd1d2d4bd044084755b93fedb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 13:33:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"95f1979c074f2b4050a0e1128d197f0661237e13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 14:02:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4611f45d35d81f93d8c0bd361ddb117195ba0a8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-14 14:32:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"af055c8c64746b89e7fe4c6f8f87f02173cebb07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 16:11:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5fb302e7f64e4baa71232879b8dc56 : SUCCESS in 9m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61ba76cac32140e491ac32889e168e33 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7be7055a01e4228b109019e2a505dbb : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21b9cd93afa34e09bdddd474ecea7880 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffcabeb237a94b92ada05209a2754ed6 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/091519694c104406bf045d63db442320 : SUCCESS in 7m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a1bbf5deceae44918db43cb34f3d11c1 : SUCCESS in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d181ff5c621d4c21bbeeccb7c0ad8ad3 : SUCCESS in 7m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/0c56ffa4da5e47eb952edaa2aa2291bf : SUCCESS in 1h 30m 20s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a96a0322c7104d548a9422b4ee6a0a71 : FAILURE in 7m 54s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0834449f80634345b289d669bd79f01d : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":6},{"id":"f878cd1ff1722be1f52e2af64657c1ea31a6503a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-20 11:18:48.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"d1efb2449fce9589da8649c4b9c3878065c68268","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-21 07:22:19.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"51f324b52c06c00d36e3a4dc13d1e1b3dcfe72b5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-21 12:15:11.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nThe patch looks good to me + unblocks gate of stable/train.","accounts_in_message":[],"_revision_number":6},{"id":"5257edd5d8efd3e3ca29c8e1fa34555ab69cd583","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 12:15:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"050b85b3e4f270b3239b25220435daf7be933676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 13:43:18.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0581fcd2074262a789c31f196c472b : SUCCESS in 8m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b24339711ab44bd7bb322d5f33a8f4ec : SUCCESS in 9m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaabea9a0d8c401fbafaaa3d23b0646c : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/078f5047bb1e4fde96ed3c7666d6cb74 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5027dbdb8f8341ef81fa1b212ab8f492 : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12f26a0c538d44cbbeb20c44965039e4 : SUCCESS in 5m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0ecca40403b34de69c3fadd0c54a901c : SUCCESS in 2m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/0e7233a8e8d04f90a825927d45da67c3 : SUCCESS in 1h 22m 38s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1f58ea944b684ec3ad98a4e6317db1f5 : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":6},{"id":"b25d115d8bfabe13cc71fa2e4f8861bb6398ce73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 13:43:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"cbf84a3bac4fd94a4f782433ddc1524e456418f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 13:56:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588ef6d039534500ba41739e0d32b992 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b9e9f54725b45b0a74457bcfca8e2b4 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f58058aa06eb5648950e9c2a6f663843570ed8cd","revisions":{"9cfc5da6ecae4d3d0a430745cad985fcda4cd57b":{"kind":"REWORK","_number":1,"created":"2021-01-12 16:31:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/50/770450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/50/770450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/1"}}},"commit":{"parents":[{"commit":"613ba308588fbb3145e6773146fef65829bfae29","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/613ba308588fbb3145e6773146fef65829bfae29"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 16:30:55.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 16:30:55.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I321dcffc1ab2494de4bc398dfc97e5b5a6666495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9cfc5da6ecae4d3d0a430745cad985fcda4cd57b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9cfc5da6ecae4d3d0a430745cad985fcda4cd57b"}]},"branch":"refs/heads/stable/train"},"68eb7f9489e925d11a4c5727dc66bfa1a4d6350b":{"kind":"REWORK","_number":2,"created":"2021-01-14 09:14:52.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/50/770450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/50/770450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/2"}}},"commit":{"parents":[{"commit":"613ba308588fbb3145e6773146fef65829bfae29","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/613ba308588fbb3145e6773146fef65829bfae29"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 16:30:55.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-14 09:14:13.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I321dcffc1ab2494de4bc398dfc97e5b5a6666495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/68eb7f9489e925d11a4c5727dc66bfa1a4d6350b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/68eb7f9489e925d11a4c5727dc66bfa1a4d6350b"}]},"branch":"refs/heads/stable/train"},"394759260a333a24ae19ded6d292db230b45ff63":{"kind":"REWORK","_number":3,"created":"2021-01-14 09:53:03.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/50/770450/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/50/770450/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/3"}}},"commit":{"parents":[{"commit":"613ba308588fbb3145e6773146fef65829bfae29","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/613ba308588fbb3145e6773146fef65829bfae29"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-14 09:52:19.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-14 09:52:19.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nFor train, add isort to test-requirements to pass tests with new pip\nresolver, and cap neutron in doc requirements to use correct version.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I321dcffc1ab2494de4bc398dfc97e5b5a6666495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/394759260a333a24ae19ded6d292db230b45ff63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/394759260a333a24ae19ded6d292db230b45ff63"}]},"branch":"refs/heads/stable/train"},"20ca63f74987c6a534d0409829512ce26a1dfedd":{"kind":"REWORK","_number":4,"created":"2021-01-14 13:33:56.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/50/770450/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/50/770450/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/4"}}},"commit":{"parents":[{"commit":"613ba308588fbb3145e6773146fef65829bfae29","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/613ba308588fbb3145e6773146fef65829bfae29"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-14 09:52:19.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-14 13:33:27.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nFor train, add isort to test-requirements to pass tests with new pip\nresolver, and use upper-constraints in doc requirements to cap neutron.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I321dcffc1ab2494de4bc398dfc97e5b5a6666495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/20ca63f74987c6a534d0409829512ce26a1dfedd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/20ca63f74987c6a534d0409829512ce26a1dfedd"}]},"branch":"refs/heads/stable/train"},"967e5b6f49b76d1d19c28085cca5e64e5ff8c07f":{"kind":"REWORK","_number":5,"created":"2021-01-14 14:02:58.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/50/770450/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/50/770450/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/5"}}},"commit":{"parents":[{"commit":"613ba308588fbb3145e6773146fef65829bfae29","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/613ba308588fbb3145e6773146fef65829bfae29"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-14 09:52:19.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-14 14:02:48.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nFor train, add isort to test-requirements to pass tests with new pip\nresolver, and use upper-constraints in doc requirements to cap neutron.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I321dcffc1ab2494de4bc398dfc97e5b5a6666495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/967e5b6f49b76d1d19c28085cca5e64e5ff8c07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/967e5b6f49b76d1d19c28085cca5e64e5ff8c07f"}]},"branch":"refs/heads/stable/train"},"f58058aa06eb5648950e9c2a6f663843570ed8cd":{"kind":"REWORK","_number":6,"created":"2021-01-14 14:32:46.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/50/770450/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/50/770450/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/770450/6"}}},"commit":{"parents":[{"commit":"613ba308588fbb3145e6773146fef65829bfae29","subject":"Merge \"bagpipe-bgp: cleanly ignore RTC route of unsupported type\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/613ba308588fbb3145e6773146fef65829bfae29"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-14 09:52:19.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-14 14:32:33.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Train)","message":"Dropping lower constraints testing (stable Train)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\nFor train, add isort to test-requirements to pass tests with new pip\nresolver, and use upper-constraints in doc requirements to cap neutron.\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I321dcffc1ab2494de4bc398dfc97e5b5a6666495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f58058aa06eb5648950e9c2a6f663843570ed8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f58058aa06eb5648950e9c2a6f663843570ed8cd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
