)]}'
{"id":"openstack%2Fnetworking-bagpipe~770452","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Frocky~I95c869f1d429f5cae168526b2beccfaf43e3d032","project":"openstack/networking-bagpipe","branch":"stable/rocky","topic":"oslo_lc_drop","hashtags":[],"change_id":"I95c869f1d429f5cae168526b2beccfaf43e3d032","subject":"Dropping lower constraints testing (stable Rocky)","status":"ABANDONED","created":"2021-01-12 16:35:13.000000000","updated":"2021-03-24 10:09:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"17485b9a266f489a6907743920c22b90832601a5","_number":770452,"virtual_id_number":770452,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-22 10:08:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 18:08:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-20 11:25:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-02-18 09:46:29.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"}],"messages":[{"id":"7d444c347cab80bc92dff5b238f463759d06ebd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-12 16:35:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"145e01caaa33c1fa6cba5f6b9269a25c4a5a9bbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 18:08:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4feee630484613ba5a5e34da647696 : FAILURE in 13m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df3488e2f99640569326a81148141edd : SUCCESS in 12m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ff487f9da34a4f1e99c64194dc87643b : SUCCESS in 10m 59s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/34ffaf61035c4663ad820488785dc6fe : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12c93813b3574c8f8de2a101f22f04f2 : SUCCESS in 3m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/acda362fddec4d43b72884bd3d742fe8 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b738f91789ee46f4b30eef5b3532fef8 : SUCCESS in 7m 12s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/4adea2a151674c49a8a9b65915f281cd : FAILURE in 8m 24s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1f5090f3fceb467fa3d05736710f7589 : SUCCESS in 1h 04m 59s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/37867332c39847c8990ef2206fc3d6b4 : FAILURE in 7m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f06ec9a9e5a3b800dffa06269fdd60c2ba7c4d5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-20 11:25:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7925879abaec4c0d2483c0b0cdabb65cce2eb9cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 12:50:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca9f763db47456cb6a02389aa106424 : FAILURE in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/160a4bf4910444f8aa295318ab942632 : SUCCESS in 8m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8372f40fba444e338781e18d7549c0f8 : SUCCESS in 8m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8a93a09c946842468a0462d9fcf755a9 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80f211ffc8384c3abecddcb7cf029430 : SUCCESS in 4m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/93252a6f22d54f5ea411d64719412e5a : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/476c58d8d585455899e24fed63ff2332 : SUCCESS in 7m 17s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/9e125147534e484188b739029e85004a : FAILURE in 11m 49s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/bb7df790601143bdaea114d48a4a863d : SUCCESS in 1h 18m 32s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/298efe24133e4c5c8ed579e042081da6 : FAILURE in 11m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad0706be9a9413218ce52fc0ea57820997384beb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-21 07:47:34.000000000","message":"Patch Set 1:\n\nstrange, pep8 works locally....","accounts_in_message":[],"_revision_number":1},{"id":"92fb847f1130fd3372eaec5b924c35cf4b114bde","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-21 07:50:58.000000000","message":"Patch Set 1:\n\nbut for tempest pip can\u0027t be found, I check how can I make sure that pip is present","accounts_in_message":[],"_revision_number":1},{"id":"4496928b0f51a2d39bfe7b60119459d6323453fb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-18 09:46:29.000000000","message":"Patch Set 1:\n\n\u003e but for tempest pip can\u0027t be found, I check how can I make sure that pip is present\nIt sounds similar to fix I needed in neutron rocky, setting  ensure_pip_from_packages_with_python2 variable:\nhttps://review.opendev.org/c/openstack/neutron/+/738851","accounts_in_message":[],"_revision_number":1},{"id":"cf2c11585dbdc56ae0577fb807e6c4b1033cb611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-18 13:29:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"187aa29bbb25ce0f3b477514a42b42a2bf542e1b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-18 13:30:22.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e but for tempest pip can\u0027t be found, I check how can I make sure that pip is present\n\u003e It sounds similar to fix I needed in neutron rocky, setting  ensure_pip_from_packages_with_python2 variable:\n\u003e https://review.opendev.org/c/openstack/neutron/+/738851\n\nThanks, let\u0027s see if it is enough","accounts_in_message":[],"_revision_number":2},{"id":"4bd852b0c11a3513c2c67482766b7cb6450d0531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 15:03:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de4a84c3a99449949c46a29afc488c48 : FAILURE in 7m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/321d7bae63584b62a577faf18461c81f : SUCCESS in 11m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/19d7ef6e15eb4d4f812c02ff214e91ec : SUCCESS in 9m 20s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0db40f17b86f477aa9f0f538ab77fa6f : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0df3654d1cf44fe0980cb2cea9607156 : SUCCESS in 6m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1899f77bf5564ee49a2cfe84b99e5ae7 : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/416d29adc5f6475889ef4abddd897014 : SUCCESS in 8m 57s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/973c397ece8b4d22b088915ed39d19f9 : SUCCESS in 54m 01s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7a3a913ef1e14f59b8233207a0b3890b : SUCCESS in 1h 24m 26s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0eb0717b19284ed18f07385d4df0a7bd : FAILURE in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"efc674e5451ba44a0f9134bff85025183456b0ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-19 11:12:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8153f4c21565c77d13fb6c96c568527b4a9a3f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 12:27:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d8d1b1b6794a13a615b996de2af3c7 : FAILURE in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd9a6d021ba3402e855001d5d1106d32 : SUCCESS in 9m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cd057ae31f5742e690f9d51fef4132b3 : FAILURE in 4m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6697ccfb0f2a41f2be06b174d9e6edac : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c423928fb2cd4af6afa163309f32799d : SUCCESS in 4m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/71c699b6ca9f4c45bf7ca67ada8de0d0 : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/211bd34c68a248ce9ee0d2c2734575a4 : SUCCESS in 7m 23s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/3bb12c8ad9b64ea48cfd117683e61db3 : SUCCESS in 47m 03s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0a9387d35a4a4e73a79a504dd6f02435 : SUCCESS in 1h 09m 16s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e1a95035c3924e05b8b716befb3d48e0 : FAILURE in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa299801a269548d4fadbf1ae876fa82d51389a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-19 13:12:23.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"df40490a39d781dd0db2cad156eb7e49dde79430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 14:22:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e70716fea94c74958a48643920697e : FAILURE in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df7737592c45497f8ba0df6cd8c90ee0 : SUCCESS in 8m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d597fcdc563d4fb2be85afbb96b52da6 : SUCCESS in 8m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a4113a3b489e4fc0808919eb7426064a : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52bd497b62b34698947c13a8e70910ab : SUCCESS in 5m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d1e4c4973fb4319aafe6a4f7d4b3859 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/504860356785458b9ad56c40c4cec3ac : SUCCESS in 8m 23s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/13ef32d8aa5241ebb4c7f90d747ea650 : SUCCESS in 35m 56s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ff072de817774f838afa236961d9a5c2 : SUCCESS in 1h 06m 06s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f53a6e1274b042bbbc13c76dcf24e387 : FAILURE in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"23fa847bf7bccd4acf1629dc3bfbb6ed58709845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-22 08:43:04.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d648632974b678d7a09e43e7864b061866ac1a75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:08:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f19152766e42e99cd35e6a1acb964c : FAILURE in 9m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae73daac75a94f5c8897eccc2b4e1a64 : SUCCESS in 8m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/57dccf28645d48ef8953a750df3a1d28 : SUCCESS in 9m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/488962d89ede4cdfa04a1c141a237dc0 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae1d5c8e767142009da70fccec31a2b9 : SUCCESS in 4m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a8dfa8caae544ba2925eb851e3e0b834 : FAILURE in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19aeb44366f1410fa335a2fef3546c3f : SUCCESS in 9m 04s\n- legacy-tempest-dsvm-networking-bagpipe https://zuul.opendev.org/t/openstack/build/235d0e7294ea4e41b091137bb87772e3 : SUCCESS in 54m 49s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2f8448d6e04e41e7af61b625184a613f : SUCCESS in 1h 18m 40s\n- legacy-networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/23acdbd7d9d24af7beb5a835158e8576 : FAILURE in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"17485b9a266f489a6907743920c22b90832601a5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-24 10:09:36.000000000","message":"Abandoned\n\nI have no time now to work on Rocky fixes.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a9ae35d82a5a0536f19252bd16cf35ecd3a16cbf","revisions":{"e0faac05bcd62e2a0ff7e880902dbd30f0f0ded3":{"kind":"REWORK","_number":1,"created":"2021-01-12 16:35:13.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/770452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/52/770452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/1"}}},"commit":{"parents":[{"commit":"dfe50a56c2c882254e332ce62220051725a970b1","subject":"Update Zuul releasenotes setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dfe50a56c2c882254e332ce62220051725a970b1"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 16:34:35.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 16:34:35.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Rocky)","message":"Dropping lower constraints testing (stable Rocky)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I95c869f1d429f5cae168526b2beccfaf43e3d032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e0faac05bcd62e2a0ff7e880902dbd30f0f0ded3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e0faac05bcd62e2a0ff7e880902dbd30f0f0ded3"}]},"branch":"refs/heads/stable/rocky"},"578dcd78a54a3f19e87f554fa5f2b02dd5781a16":{"kind":"REWORK","_number":2,"created":"2021-02-18 13:29:26.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/770452/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/52/770452/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/2"}}},"commit":{"parents":[{"commit":"dfe50a56c2c882254e332ce62220051725a970b1","subject":"Update Zuul releasenotes setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dfe50a56c2c882254e332ce62220051725a970b1"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 16:34:35.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-02-18 13:28:39.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Rocky)","message":"Dropping lower constraints testing (stable Rocky)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I95c869f1d429f5cae168526b2beccfaf43e3d032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/578dcd78a54a3f19e87f554fa5f2b02dd5781a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/578dcd78a54a3f19e87f554fa5f2b02dd5781a16"}]},"branch":"refs/heads/stable/rocky"},"3689b43b1b0e81f69b074147fc136ab75be32982":{"kind":"REWORK","_number":3,"created":"2021-02-19 11:12:28.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/770452/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/52/770452/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/3"}}},"commit":{"parents":[{"commit":"dfe50a56c2c882254e332ce62220051725a970b1","subject":"Update Zuul releasenotes setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dfe50a56c2c882254e332ce62220051725a970b1"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 16:34:35.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-19 11:11:45.000000000","tz":60},"subject":"Dropping lower constraints testing (stable Rocky)","message":"Dropping lower constraints testing (stable Rocky)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I95c869f1d429f5cae168526b2beccfaf43e3d032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3689b43b1b0e81f69b074147fc136ab75be32982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3689b43b1b0e81f69b074147fc136ab75be32982"}]},"branch":"refs/heads/stable/rocky"},"9ce22a6845bed726456f4ffaa6661f99aa55d021":{"kind":"REWORK","_number":4,"created":"2021-02-19 13:12:23.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/770452/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/52/770452/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/4"}}},"commit":{"parents":[{"commit":"dfe50a56c2c882254e332ce62220051725a970b1","subject":"Update Zuul releasenotes setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dfe50a56c2c882254e332ce62220051725a970b1"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 16:34:35.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-02-19 13:11:18.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Rocky)","message":"Dropping lower constraints testing (stable Rocky)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I95c869f1d429f5cae168526b2beccfaf43e3d032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9ce22a6845bed726456f4ffaa6661f99aa55d021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9ce22a6845bed726456f4ffaa6661f99aa55d021"}]},"branch":"refs/heads/stable/rocky"},"a9ae35d82a5a0536f19252bd16cf35ecd3a16cbf":{"kind":"REWORK","_number":5,"created":"2021-02-22 08:43:04.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/770452/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/52/770452/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/52/770452/5"}}},"commit":{"parents":[{"commit":"dfe50a56c2c882254e332ce62220051725a970b1","subject":"Update Zuul releasenotes setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dfe50a56c2c882254e332ce62220051725a970b1"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-12 16:34:35.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-02-22 08:43:01.000000000","tz":0},"subject":"Dropping lower constraints testing (stable Rocky)","message":"Dropping lower constraints testing (stable Rocky)\n\nThis topic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nAs decided in Neutron meeting, stable branches will drop this\nCI job [2].\n\n[1]http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n[2]http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html\n\nChange-Id: I95c869f1d429f5cae168526b2beccfaf43e3d032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a9ae35d82a5a0536f19252bd16cf35ecd3a16cbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a9ae35d82a5a0536f19252bd16cf35ecd3a16cbf"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
