)]}'
{"id":"openstack%2Fnetworking-bagpipe~775398","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ia5f4d33421f12ffacf88764da8a1bc4659c4a398","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/1915522","hashtags":[],"change_id":"Ia5f4d33421f12ffacf88764da8a1bc4659c4a398","subject":"Flood broadcast to dhcp tunnels","status":"ABANDONED","created":"2021-02-12 14:50:35.000000000","updated":"2023-02-28 11:13:21.000000000","total_comment_count":8,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"bfa8ed1355950f3577d6bba36b5da1c7a9c88b4e","_number":775398,"virtual_id_number":775398,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-08 15:22:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":2,"date":"2021-03-09 11:48:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-03-09 14:06:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2021-03-09 12:25:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":-1,"date":"2021-03-12 05:11:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-12 16:29:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-02-16 12:37:54.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-03 09:45:42.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REMOVED"}],"messages":[{"id":"377426ce56333a3ed5b959a02c12c8c2df21d858","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-02-12 14:50:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6e4a2c4e951437b17e3390b87735fff4ef044e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 16:29:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606b9269b94c40c6bac13c24e0b4d932 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ea8873d15454be29bc6ff4510012f09 : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41ddc82a5649413b870882d7a726ffcc : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/142a3eaf97cd46318bb1172643003ea6 : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d295fb5ac3af47ccb80f7ac34e1042be : SUCCESS in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc2377cfabb748f9824b56f9ac686b94 : FAILURE in 4m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6331627f27a54f83ba912f5d0aab3331 : SUCCESS in 8m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1380cd1d0b2b412588eafc8df5a6ab14 : FAILURE in 1h 36m 01s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/c7e941f1486c4778931f748708837196 : FAILURE in 15m 55s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/06753508f89c41b0b61c529b6f28564e : SUCCESS in 1h 31m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d6d12b496c3782ac6784cb650f096596aef90a35","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-02-12 21:20:32.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57a2aaab0a01ad4d3609e182d7f4959d018e60e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-02-12 21:40:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2ba0d935c7ef22c09d4da3cbebe397adbd62a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 23:25:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c002d1a65c94480ac7d38a13a4d558a : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55be42f11aa84b75807566dc9b5d3bf3 : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a8a735d22004d799dd0b0ea42c96264 : SUCCESS in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a2a3b28b2d74880bd5574e02f24d56c : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/185aeadd28914aebb481babc60cb06b3 : SUCCESS in 6m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d1ead35b42e4c26926e4e80b03fddf2 : FAILURE in 4m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fde00a56d8d475ca25a083772edd0b9 : SUCCESS in 12m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/53f59a7053734e3e8183b0b69f91cac1 : FAILURE in 1h 38m 05s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/769ce31ac75d47e5a9fe49e16aeb5956 : FAILURE in 16m 32s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1107413b980c487caad43daf88f246a7 : SUCCESS in 1h 38m 10s","accounts_in_message":[],"_revision_number":2},{"id":"818316f0aa906c995010b09d73b72969476dc5fd","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-02-14 07:41:08.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d2b14aada11aba15e0eca1c1e7511a88390b4559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-02-15 08:36:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c7bcaa9259f65961a382c193f3a4fe3704f1c10a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-02-15 09:16:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7516fa6e611f53468ef3cb1abde165bb2df54f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 11:01:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f504e1d863ce47a6b7d629de9d7b7295 : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1a52551adb44b198f9646055377517c : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26b99350969a4cd99679e11796b24f1a : SUCCESS in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/105a092d606d4c51891332d59ba3d445 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79179ee7517f42739a4e7f1344046069 : SUCCESS in 7m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d5c04f6a4de4a58a48e8dbec0dcd639 : FAILURE in 4m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/400f2023fdf94797935683325973ac27 : SUCCESS in 9m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/65fd9f230fe343e4a55b6fcda97b4fbd : FAILURE in 1h 36m 02s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/d5bbe55668e24a4d85fcdc3beadf36c9 : FAILURE in 17m 23s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/7ed4ac034945476599cd43031111358c : SUCCESS in 1h 38m 22s","accounts_in_message":[],"_revision_number":4},{"id":"8bad236e45adff85709078be2e0a19c18d20fc4b","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-02-15 18:43:58.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"088149f83c4ec772a5f5484ee37b415d74536c11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-02-16 09:41:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"31f5a8226e168c857a2f30713c073f338de12d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 11:23:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0856a6f74dec41459aed051581ce067b : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/445a577978f04c5ea66d16654ef700ed : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c0ade2139814a56921c179e5aad777f : SUCCESS in 8m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f09c64514e940e5aa6e291882c0d276 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3c92026d454d4582343e3d72e68086 : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b48f9febf1b74baa91e9a0a42768387a : FAILURE in 5m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59365d88472e477a9e64dd6a3a394812 : SUCCESS in 9m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ba9586dfa78544d1b84dcd4a80963c01 : SUCCESS in 1h 31m 46s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2afd055d61bd42c4b34cb56d87a7f7ef : FAILURE in 14m 18s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/8e64ae0fbc174ba2b4e472bf3b01048a : SUCCESS in 1h 34m 46s","accounts_in_message":[],"_revision_number":5},{"id":"3ffd51177d4c135e59056c885cc026b5ca9fa1fe","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-02-16 12:37:54.000000000","message":"Patch Set 5:\n\nplease review, thanks in advance","accounts_in_message":[],"_revision_number":5},{"id":"72442e1c5584e0c4540fed71f287b6769da88972","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-02-24 13:19:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"983419c22dbd8f0e4ff4e761a5cfd5b0f0227e52","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-26 11:08:35.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nI am fine with it please fix the nits from Liu adn from me.","accounts_in_message":[],"_revision_number":5},{"id":"4032c3e7e4528ae10e63cd4b1d87321fd0e78d7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-02 21:17:38.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"ee0f8c5eab7b68903c5803044f93e24daa34c7a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-02 21:19:28.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"cea758dc22d299eee5cfba276ead00ab25236916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 22:47:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/834f0238feab470e96f9530adc384681 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b172353cbe814820aa729d3e733d1dd1 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee6c7fd2a4874adc8a2e9411389428d0 : SUCCESS in 8m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0141150f628487187bea4a3dcf9cac7 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4ed1daff2a4c33bf4c0625f0579a02 : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/276399ae9b834778a6f76b3fa236caa2 : FAILURE in 4m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/200264c7c47c4da3b8d86c121c7bc2eb : SUCCESS in 10m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7a1d18aaddee4698bb3e7bfca0f7920c : SUCCESS in 1h 11m 50s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/9a638de30b574228b68f1d5f14ed4515 : FAILURE in 14m 08s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/d12f46d3a0c5497bb53fb3868153b28e : SUCCESS in 1h 21m 43s","accounts_in_message":[],"_revision_number":7},{"id":"2b371b90ef2accc449b01c84aad3b3bbc341b837","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-03 05:46:14.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"d6aa5bfe6e8550689e8218e35f0a23dd4a31538e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-03 05:51:22.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"d71e86aa91f00137b38f0bd82a4fe3c1a9d36a00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 07:27:30.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca511b8bcd14f4d932009ce1a7d5892 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c167f7726e4b4d05aaa09b6adcc890bf : SUCCESS in 12m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34fb020a315d4cd59e49cb23704065ea : SUCCESS in 8m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e72365a58025487ba2ea65fbd805e953 : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05408870238c4316b82a5e370445aea0 : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdef908aefbf46bd85a5dd0bc50a1174 : FAILURE in 4m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf422adf5da943db8a19b84486177d52 : SUCCESS in 7m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/94f7790bea284acaa78b3b414d577bd8 : SUCCESS in 1h 11m 37s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/361f8991b57842b4a06f83d3f1560525 : FAILURE in 16m 57s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/2a9759b2a8274f098f78b3c518f9e9bd : SUCCESS in 1h 33m 21s","accounts_in_message":[],"_revision_number":8},{"id":"c3998474e4fbf92aa7beeaee27db3e0ed3622b7f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-03 08:48:33.000000000","message":"Patch Set 8: Code-Review+2\n\nThanks for working on this","accounts_in_message":[],"_revision_number":8},{"id":"1ee1863ea64a182c640c8d81c9707a5e62c983c9","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-03-03 09:45:42.000000000","message":"Removed reviewer Lucas Alvares Gomes.","accounts_in_message":[],"_revision_number":8},{"id":"26ad44fe2ed3194c17cdb33ba1d7e7e71683bc5a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-04 04:02:27.000000000","message":"Topic mstr_fix removed","accounts_in_message":[],"_revision_number":8},{"id":"da97098d1971c02f8a2f39cd7537c35b0f05718c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-04 04:02:34.000000000","message":"Topic set to bug/1915522","accounts_in_message":[],"_revision_number":8},{"id":"9aa84c5fee0887644d83ea42ac5bffe60a7f1130","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-04 04:02:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ee609470cbc365fe1a215c9b725fd8268e6be59a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-05 07:45:00.000000000","message":"Patch Set 8: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b5dd8d499ace2b685304315289542baf8a63ba7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-08 13:20:59.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"cd1d0d09766488551d5c5f7ca68e3293b6f541f8","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-08 13:26:02.000000000","message":"Patch Set 9:\n\nLajos, Akihiro thanks for review, uploaded patch to neutron https://review.opendev.org/c/openstack/neutron/+/779149/2, added todo in next PS.","accounts_in_message":[],"_revision_number":9},{"id":"e642801e63a595e848ca8f2bf8d4bb9cb237e505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:22:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1792d8d9e6ac49888daf449988f60489 : SUCCESS in 10m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39d321e916fb4896966b3a93b1a596a3 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e34cde4fc77d465290a3049aadb826b0 : SUCCESS in 9m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30e20da5279c4816845b8e596359ee50 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bed5b5e641594fce9afca79bf96bba32 : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9addb4c2233346e4be35faa2af4598dd : FAILURE in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bece61238e1d43409bb984e4b42917ba : SUCCESS in 11m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0ed426f22d524d6e938c64e26f28cb6b : SUCCESS in 1h 18m 02s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/3f39b563981d4ab7be3ae0921acae64f : FAILURE in 13m 00s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/5ee97bf770d24e85a8df4f5d7465616e : SUCCESS in 1h 53m 14s","accounts_in_message":[],"_revision_number":9},{"id":"1ef295d0e866ba83b95b7af552db948357074df1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-09 11:48:44.000000000","message":"Patch Set 9: Code-Review+2\n\nLooks ok for me but someone with more experience in bagpipe needs to check it","accounts_in_message":[],"_revision_number":9},{"id":"512c0b08a59759f1d27b5613bac06cbd78e6bf70","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-09 12:25:14.000000000","message":"Patch Set 9: Code-Review+2\n\nI try to ping tmorin as well for review","accounts_in_message":[],"_revision_number":9},{"id":"3d30e5b1a5d2199d86799676739e77c09c7236fd","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-09 14:06:26.000000000","message":"Patch Set 9: Code-Review+1\n\nLGTM but I am not familiar with this area so just +1.","accounts_in_message":[],"_revision_number":9},{"id":"efc365cb137bb96227b9d5e90e31afb78a117611","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-03-12 05:11:09.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bfa8ed1355950f3577d6bba36b5da1c7a9c88b4e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 11:13:21.000000000","message":"Abandoned\n\nIf needed, feel free to restore the patch and fix the merge conflict","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"2ce59564df3adcc5359d8fca81003c73850c1572","revisions":{"399e43c1939fc92d7d6c7db67ca804299ecb0d72":{"kind":"REWORK","_number":1,"created":"2021-02-12 14:50:35.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/98/775398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/775398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/1"}}},"commit":{"parents":[{"commit":"0146eda9d915e53dffb2107dd12c2ffa15e90e7c","subject":"Make ovs evpn driver an os-ken app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0146eda9d915e53dffb2107dd12c2ffa15e90e7c"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 13:20:17.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 14:49:24.000000000","tz":120},"subject":"Flood broadcast to dhcp tunnels","message":"Flood broadcast to dhcp tunnels\n\nThe patch makes sure broadcast traffic is sent not only to EVPN\ntunnels, but for original neutron tunnels with DHCP).\n\nCloses-Bug: #1915522\nChange-Id: Ia5f4d33421f12ffacf88764da8a1bc4659c4a398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/399e43c1939fc92d7d6c7db67ca804299ecb0d72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/399e43c1939fc92d7d6c7db67ca804299ecb0d72"}]},"branch":"refs/heads/master"},"3ae2d9c08cfd4f9407f39d1fc2d5e189bce73b95":{"kind":"REWORK","_number":2,"created":"2021-02-12 21:40:53.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/98/775398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/775398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/2"}}},"commit":{"parents":[{"commit":"0146eda9d915e53dffb2107dd12c2ffa15e90e7c","subject":"Make ovs evpn driver an os-ken app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0146eda9d915e53dffb2107dd12c2ffa15e90e7c"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 13:20:17.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 21:39:02.000000000","tz":120},"subject":"Flood broadcast to dhcp tunnels","message":"Flood broadcast to dhcp tunnels\n\nThe patch makes sure broadcast traffic is sent not only to EVPN\ntunnels, but for original neutron tunnels with DHCP).\n\nCloses-Bug: #1915522\nChange-Id: Ia5f4d33421f12ffacf88764da8a1bc4659c4a398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3ae2d9c08cfd4f9407f39d1fc2d5e189bce73b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3ae2d9c08cfd4f9407f39d1fc2d5e189bce73b95"}]},"branch":"refs/heads/master"},"b81d0957d356b1af709d085f0039668c03f41f5a":{"kind":"REWORK","_number":3,"created":"2021-02-15 08:36:08.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/98/775398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/775398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/3"}}},"commit":{"parents":[{"commit":"0146eda9d915e53dffb2107dd12c2ffa15e90e7c","subject":"Make ovs evpn driver an os-ken app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0146eda9d915e53dffb2107dd12c2ffa15e90e7c"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 13:20:17.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-15 08:35:52.000000000","tz":120},"subject":"Flood broadcast to dhcp tunnels","message":"Flood broadcast to dhcp tunnels\n\nThe patch makes sure broadcast traffic is sent not only to EVPN\ntunnels, but for original neutron tunnels with DHCP).\n\nCloses-Bug: #1915522\nChange-Id: Ia5f4d33421f12ffacf88764da8a1bc4659c4a398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b81d0957d356b1af709d085f0039668c03f41f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b81d0957d356b1af709d085f0039668c03f41f5a"}]},"branch":"refs/heads/master"},"df87f4b11883a604fa0eb85200e78cf17280c39e":{"kind":"REWORK","_number":4,"created":"2021-02-15 09:16:20.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/98/775398/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/775398/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/4"}}},"commit":{"parents":[{"commit":"0146eda9d915e53dffb2107dd12c2ffa15e90e7c","subject":"Make ovs evpn driver an os-ken app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0146eda9d915e53dffb2107dd12c2ffa15e90e7c"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 13:20:17.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-15 09:15:56.000000000","tz":120},"subject":"Flood broadcast to dhcp tunnels","message":"Flood broadcast to dhcp tunnels\n\nThe patch makes sure broadcast traffic is sent not only to EVPN\ntunnels, but for original neutron tunnels with DHCP).\n\nCloses-Bug: #1915522\nChange-Id: Ia5f4d33421f12ffacf88764da8a1bc4659c4a398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/df87f4b11883a604fa0eb85200e78cf17280c39e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/df87f4b11883a604fa0eb85200e78cf17280c39e"}]},"branch":"refs/heads/master"},"899f7c2abf5256a48bb9acb247f12990db03df64":{"kind":"REWORK","_number":5,"created":"2021-02-16 09:41:08.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/98/775398/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/775398/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/5"}}},"commit":{"parents":[{"commit":"0146eda9d915e53dffb2107dd12c2ffa15e90e7c","subject":"Make ovs evpn driver an os-ken app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0146eda9d915e53dffb2107dd12c2ffa15e90e7c"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 13:20:17.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-16 09:40:16.000000000","tz":120},"subject":"Flood broadcast to dhcp tunnels","message":"Flood broadcast to dhcp tunnels\n\nThe patch makes sure broadcast traffic is sent not only to EVPN\ntunnels, but for original neutron tunnels with DHCP). Mirror\nrule for ARP responder with higher priority than flooding.\n\nCloses-Bug: #1915522\nChange-Id: Ia5f4d33421f12ffacf88764da8a1bc4659c4a398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/899f7c2abf5256a48bb9acb247f12990db03df64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/899f7c2abf5256a48bb9acb247f12990db03df64"}]},"branch":"refs/heads/master"},"e93ab77b2bda8b403ac5982e53f090df29f69ded":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-03-02 21:17:38.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/98/775398/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/775398/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/6"}}},"commit":{"parents":[{"commit":"3a20f0fb7f47c9e203ef6db8c4a39e90a8299d38","subject":"Make ovs evpn driver an os-ken app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3a20f0fb7f47c9e203ef6db8c4a39e90a8299d38"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 13:20:17.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-02 21:16:56.000000000","tz":120},"subject":"Flood broadcast to dhcp tunnels","message":"Flood broadcast to dhcp tunnels\n\nThe patch makes sure broadcast traffic is sent not only to EVPN\ntunnels, but for original neutron tunnels with DHCP). Mirror\nrule for ARP responder with higher priority than flooding.\n\nCloses-Bug: #1915522\nChange-Id: Ia5f4d33421f12ffacf88764da8a1bc4659c4a398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e93ab77b2bda8b403ac5982e53f090df29f69ded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e93ab77b2bda8b403ac5982e53f090df29f69ded"}]},"branch":"refs/heads/master"},"fb5469f5d15a3079bdb3bec8da9854dd5150617b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-03-02 21:19:28.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/98/775398/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/775398/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/7"}}},"commit":{"parents":[{"commit":"4fb14bf629de2acd20e982816167ab12e2c47ee4","subject":"Make ovs evpn driver an os-ken app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4fb14bf629de2acd20e982816167ab12e2c47ee4"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 13:20:17.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-02 21:19:28.000000000","tz":0},"subject":"Flood broadcast to dhcp tunnels","message":"Flood broadcast to dhcp tunnels\n\nThe patch makes sure broadcast traffic is sent not only to EVPN\ntunnels, but for original neutron tunnels with DHCP). Mirror\nrule for ARP responder with higher priority than flooding.\n\nCloses-Bug: #1915522\nChange-Id: Ia5f4d33421f12ffacf88764da8a1bc4659c4a398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fb5469f5d15a3079bdb3bec8da9854dd5150617b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fb5469f5d15a3079bdb3bec8da9854dd5150617b"}]},"branch":"refs/heads/master","description":"Rebase"},"ab1584b29b43c51a7c880126dbe3d8f7c471b086":{"kind":"REWORK","_number":8,"created":"2021-03-03 05:51:22.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/98/775398/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/775398/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/8"}}},"commit":{"parents":[{"commit":"4fb14bf629de2acd20e982816167ab12e2c47ee4","subject":"Make ovs evpn driver an os-ken app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4fb14bf629de2acd20e982816167ab12e2c47ee4"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 13:20:17.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-03 05:51:17.000000000","tz":0},"subject":"Flood broadcast to dhcp tunnels","message":"Flood broadcast to dhcp tunnels\n\nThe patch makes sure broadcast traffic is sent not only to EVPN\ntunnels, but for original neutron tunnels with DHCP. Mirror\nrule for ARP responder with higher priority than flooding.\n\nCloses-Bug: #1915522\nChange-Id: Ia5f4d33421f12ffacf88764da8a1bc4659c4a398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ab1584b29b43c51a7c880126dbe3d8f7c471b086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ab1584b29b43c51a7c880126dbe3d8f7c471b086"}]},"branch":"refs/heads/master"},"2ce59564df3adcc5359d8fca81003c73850c1572":{"kind":"REWORK","_number":9,"created":"2021-03-08 13:20:59.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/98/775398/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/775398/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/775398/9"}}},"commit":{"parents":[{"commit":"4fb14bf629de2acd20e982816167ab12e2c47ee4","subject":"Make ovs evpn driver an os-ken app","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4fb14bf629de2acd20e982816167ab12e2c47ee4"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-02-12 13:20:17.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-08 13:20:55.000000000","tz":0},"subject":"Flood broadcast to dhcp tunnels","message":"Flood broadcast to dhcp tunnels\n\nThe patch makes sure broadcast traffic is sent not only to EVPN\ntunnels, but for original neutron tunnels with DHCP. Mirror\nrule for ARP responder with higher priority than flooding.\n\nCloses-Bug: #1915522\nChange-Id: Ia5f4d33421f12ffacf88764da8a1bc4659c4a398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2ce59564df3adcc5359d8fca81003c73850c1572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2ce59564df3adcc5359d8fca81003c73850c1572"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
