)]}'
{"id":"openstack%2Fnetworking-bagpipe~778330","triplet_id":"openstack%2Fnetworking-bagpipe~master~I7369849754778c61fcf207f1823233e232d09e97","project":"openstack/networking-bagpipe","branch":"master","hashtags":[],"change_id":"I7369849754778c61fcf207f1823233e232d09e97","subject":"Handle network update notifications","status":"ABANDONED","created":"2021-03-02 21:17:38.000000000","updated":"2023-02-28 11:13:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e2c7f05e2f009f184bb7eece0a4676ca44e4636f","_number":778330,"virtual_id_number":778330,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-08 15:16:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2021-03-29 11:16:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 22:42:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 11:16:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-29 11:16:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2021-04-08 03:19:35.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"}],"messages":[{"id":"a70888b3442135988b378fb3e782222144bb6cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-02 21:17:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64491132d4b82a5f422a9fa6c374f3ae73fca661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-02 21:19:03.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"15c36ef375a6c9127ec4248941ad469de25dccfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-02 21:19:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6137344d5b4f418c6a16ee419903698aacafb86e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 22:42:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ff6c9b1ac743379bc06a69eda57176 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54dd976204fb478dbf3cefbe46cfe6d7 : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f964ffa0d4434d6e813c56da81777cdb : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dc7f5b0d48e41f088fc7cdb7f359c19 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04732979e67425f9012eb0c7d7a9d41 : SUCCESS in 7m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a73e184cb08e4caf8ac80c530bff3804 : FAILURE in 4m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/247b53fd893b4a13bc8278b188cb6af6 : SUCCESS in 9m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/078b2eb9dada4463812e6acc836bcc6e : SUCCESS in 1h 12m 33s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/32b4e3dd9e5e44d49c32d7617f7278a4 : FAILURE in 17m 18s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c9cd7a09e46a42fcb6c56d1dbb038ba0 : SUCCESS in 1h 19m 13s","accounts_in_message":[],"_revision_number":3},{"id":"ded99d5ae1c3ffb87903e94160de19fb779a88cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-03 05:51:33.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"97474ce58614d6d13aeed258ed03bec1f39aa5a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 07:29:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f88f657af2e4ea1b00804611409627c : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e3d840cd9f3442aa59ba65741e1baa9 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85ea24aab5224a6da0babd22b215318d : SUCCESS in 9m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dabc47a9754451d8db2eb6bf3835811 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e74ac9d350f4fafab2b58c90d6a5fc7 : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2578b9e09d4c431bb60f37413dfb6e0f : FAILURE in 4m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/452a4f7496784a5f9f0679453283945a : SUCCESS in 8m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c09dc592652b42868313d20e4f15c9a2 : SUCCESS in 1h 07m 41s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/78b943283ed34dc3ae87b72a89eadd93 : FAILURE in 13m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3f8605eff67247deb03405d8dedd5848 : SUCCESS in 1h 35m 25s","accounts_in_message":[],"_revision_number":4},{"id":"a16b9fe6cff5d8d1ec0d5bdf25a985bd8579ca2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-03-08 13:26:09.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"3ad3fa54f14846a5da1e4b1f2b78371794481b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:16:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543245f0132f4ade98543d332bc517cd : SUCCESS in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e643e172efa041d0be4f4aec92832ced : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9056b177ab9840ca8e873840f268b9d5 : SUCCESS in 8m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9351a5bdaa524cabba6041d36fdf2742 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0200e014ec1746d6a002f69013cba4be : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54cc959e616d4e829837c662d070ce32 : FAILURE in 4m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b84706484f4440a6970d6760b7e94ea7 : SUCCESS in 10m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/58aacace0d4448be9f688a74233a0228 : SUCCESS in 1h 31m 03s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/5cc94ee311e34c97966d0e6cf88874fc : FAILURE in 19m 22s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/a17c73119e98488c8fd09af41ac8508f : SUCCESS in 1h 42m 21s","accounts_in_message":[],"_revision_number":5},{"id":"eb75781e752ee9cff5375f5417735be0ed710bff","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-29 11:16:18.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"9a65d3322aafbc54d3eca4f39bfefbce9e56b29a","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-04-08 03:19:35.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e9fd6f9b8478666bbabc69af3e02aa8d0615f2fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 04:55:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b97dad99f59a4cbcad0b449e40cdd208 : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5449d41cab854f1cbbc0ab8fe7913f7f : SUCCESS in 18m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d42be0d90b04ee49564e5f729f8adfc : SUCCESS in 9m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15e02124d0f7429cabbab796ef229136 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd5b38e62c64c8195a7c3c9aba32dce : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc011a28a8d54b4fb0ada200b9489344 : FAILURE in 7m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4726fed42b0940b485390da5c7f10b44 : SUCCESS in 9m 29s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d06ef37b31954ebfa48ea4228d513e7e : SUCCESS in 1h 22m 03s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/c87ce763190c42e791ae621da6ff7377 : FAILURE in 15m 12s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1859a6f36d2c4c6f8255e90faf482064 : SUCCESS in 1h 21m 53s","accounts_in_message":[],"_revision_number":5},{"id":"e2c7f05e2f009f184bb7eece0a4676ca44e4636f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 11:13:00.000000000","message":"Abandoned\n\nIf needed, feel free to restore the patch and fix the merge conflict","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"deaa4b3810ce7918b37a75cb0e9291ea85bd1a36","revisions":{"9847144bc04b8d92dca3850207a1c7ae34a7bad4":{"kind":"REWORK","_number":1,"created":"2021-03-02 21:17:38.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/30/778330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/30/778330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/1"}}},"commit":{"parents":[{"commit":"e93ab77b2bda8b403ac5982e53f090df29f69ded","subject":"Flood broadcast to dhcp tunnels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e93ab77b2bda8b403ac5982e53f090df29f69ded"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-02 13:20:30.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-02 21:17:17.000000000","tz":120},"subject":"Handle network update notifications","message":"Handle network update notifications\n\nHandle network update notifications to make sure\nwe adjust rules for network that was plugged to router\nafter it is associated with bgpvpn.\n\nCloses-Bug: #1897408\n\nRelated-Prod: PRODX-12049\nChange-Id: I7369849754778c61fcf207f1823233e232d09e97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9847144bc04b8d92dca3850207a1c7ae34a7bad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9847144bc04b8d92dca3850207a1c7ae34a7bad4"}]},"branch":"refs/heads/master"},"0541000be7a2c8728e32859559640b40ca683ae1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-02 21:19:03.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/30/778330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/30/778330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/2"}}},"commit":{"parents":[{"commit":"e93ab77b2bda8b403ac5982e53f090df29f69ded","subject":"Flood broadcast to dhcp tunnels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e93ab77b2bda8b403ac5982e53f090df29f69ded"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-02 13:20:30.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-02 21:19:00.000000000","tz":0},"subject":"Handle network update notifications","message":"Handle network update notifications\n\nHandle network update notifications to make sure\nwe adjust rules for network that was plugged to router\nafter it is associated with bgpvpn.\n\nCloses-Bug: #1897408\n\nChange-Id: I7369849754778c61fcf207f1823233e232d09e97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0541000be7a2c8728e32859559640b40ca683ae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0541000be7a2c8728e32859559640b40ca683ae1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"490a055c0cb221fe7f372b06eeeb61ab3788387c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-02 21:19:57.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/30/778330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/30/778330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/3"}}},"commit":{"parents":[{"commit":"fb5469f5d15a3079bdb3bec8da9854dd5150617b","subject":"Flood broadcast to dhcp tunnels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fb5469f5d15a3079bdb3bec8da9854dd5150617b"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-02 13:20:30.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-02 21:19:57.000000000","tz":0},"subject":"Handle network update notifications","message":"Handle network update notifications\n\nHandle network update notifications to make sure\nwe adjust rules for network that was plugged to router\nafter it is associated with bgpvpn.\n\nCloses-Bug: #1897408\n\nChange-Id: I7369849754778c61fcf207f1823233e232d09e97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/490a055c0cb221fe7f372b06eeeb61ab3788387c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/490a055c0cb221fe7f372b06eeeb61ab3788387c"}]},"branch":"refs/heads/master","description":"Rebase"},"ae6e32ffb07235ba924dff2df540ccb1f6e4ccce":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-03 05:51:33.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/30/778330/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/30/778330/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/4"}}},"commit":{"parents":[{"commit":"ab1584b29b43c51a7c880126dbe3d8f7c471b086","subject":"Flood broadcast to dhcp tunnels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ab1584b29b43c51a7c880126dbe3d8f7c471b086"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-02 13:20:30.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-03 05:51:33.000000000","tz":0},"subject":"Handle network update notifications","message":"Handle network update notifications\n\nHandle network update notifications to make sure\nwe adjust rules for network that was plugged to router\nafter it is associated with bgpvpn.\n\nCloses-Bug: #1897408\n\nChange-Id: I7369849754778c61fcf207f1823233e232d09e97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ae6e32ffb07235ba924dff2df540ccb1f6e4ccce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ae6e32ffb07235ba924dff2df540ccb1f6e4ccce"}]},"branch":"refs/heads/master","description":"Rebase"},"deaa4b3810ce7918b37a75cb0e9291ea85bd1a36":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-03-08 13:26:09.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/30/778330/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/30/778330/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/778330/5"}}},"commit":{"parents":[{"commit":"2ce59564df3adcc5359d8fca81003c73850c1572","subject":"Flood broadcast to dhcp tunnels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2ce59564df3adcc5359d8fca81003c73850c1572"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-02 13:20:30.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2021-03-08 13:26:09.000000000","tz":0},"subject":"Handle network update notifications","message":"Handle network update notifications\n\nHandle network update notifications to make sure\nwe adjust rules for network that was plugged to router\nafter it is associated with bgpvpn.\n\nCloses-Bug: #1897408\n\nChange-Id: I7369849754778c61fcf207f1823233e232d09e97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/deaa4b3810ce7918b37a75cb0e9291ea85bd1a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/deaa4b3810ce7918b37a75cb0e9291ea85bd1a36"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
