)]}'
{"id":"openstack%2Fnetworking-bagpipe~782805","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fwallaby~I751f47e99f474069f6644a58323ece93567ca4e9","project":"openstack/networking-bagpipe","branch":"stable/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"I751f47e99f474069f6644a58323ece93567ca4e9","subject":"Update .gitreview for stable/wallaby","status":"MERGED","created":"2021-03-24 15:58:38.000000000","updated":"2021-03-26 12:05:56.000000000","submitted":"2021-03-26 12:04:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782805-create-wallaby","meta_rev_id":"9fd30a22b50f63aabad9d7a731251594cec19296","_number":782805,"virtual_id_number":782805,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 12:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-26 11:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-24 16:29:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-03-25 07:07:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-03-26 11:30:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 16:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-24 16:29:19.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-25 07:07:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-26 11:30:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"310a8ec0301864141d66dd81aa99ed39918a3ada","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-24 15:58:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09078d70a6a697ab54b7f8a13fd2ccaac1a418c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 16:28:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f1652e3655d479cbb97a494f6a2acf0 : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79fbfa0b9dbb469c88ed4d4f08e97365 : SUCCESS in 11m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d2146e048cf441980e6da676a2a07fb : SUCCESS in 11m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60782014c14d4bcba6eca1ab91fc417f : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f760276259894971b48bd69a2cf57122 : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7e6bd53b6f646ac993cb99ccbe22dce : FAILURE in 8m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f0a8ab9d0e148a98cf4abdf3598d320 : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0dcd6131d335352563111981881847f7897e38ff","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-24 16:29:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea0af6f7b0adf7e59b4f42653f17edac09261c6d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-25 07:07:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"84cc6bcadcf648b2092b2f249b24e8493e645b58","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-26 11:30:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be246fc2b69f467588689a6824241c4e3b676bda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 11:46:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf241386e5bb4040f56156bdd8d4f8defdcd08b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 12:04:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/854aec6605fb48e5a6ec10ce1b373afd : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3b92b3edad74cbf9be497739d5edc71 : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93131a58c4eb43bc973bb4321473d6b6 : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48b5084981f4815af678a6fbbc1fa95 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fa8f7784e5467540f544edbf5171d619a618f5ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 12:04:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9fd30a22b50f63aabad9d7a731251594cec19296","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 12:05:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb395456b9d48cd8872aa2b3f85fa75 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0d2f86bf85c5a8b90adb7f9df88e214eac0227a","revisions":{"e0d2f86bf85c5a8b90adb7f9df88e214eac0227a":{"kind":"REWORK","_number":1,"created":"2021-03-24 15:58:38.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/05/782805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/05/782805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/782805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/782805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/782805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/05/782805/1"}}},"commit":{"parents":[{"commit":"a7e479f043d31cbfe57f5b0039757fa8e3546547","subject":"Merge \"Set secure mode for mpls bridge\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a7e479f043d31cbfe57f5b0039757fa8e3546547"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-24 15:58:35.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-24 15:58:35.000000000","tz":0},"subject":"Update .gitreview for stable/wallaby","message":"Update .gitreview for stable/wallaby\n\nChange-Id: I751f47e99f474069f6644a58323ece93567ca4e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e0d2f86bf85c5a8b90adb7f9df88e214eac0227a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e0d2f86bf85c5a8b90adb7f9df88e214eac0227a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
