)]}'
{"id":"openstack%2Fnetworking-bagpipe~782806","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fwallaby~Ibdf843347ab05abccdd3d2745d378c603313ca18","project":"openstack/networking-bagpipe","branch":"stable/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"Ibdf843347ab05abccdd3d2745d378c603313ca18","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","status":"MERGED","created":"2021-03-24 15:58:44.000000000","updated":"2021-03-26 12:12:13.000000000","submitted":"2021-03-26 12:08:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782806-create-wallaby","meta_rev_id":"289f74baaf1273c7cd5c6c3a2a2f76b4d011a639","_number":782806,"virtual_id_number":782806,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 12:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-26 11:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-24 16:29:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-03-25 07:07:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-03-26 11:31:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 16:29:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-24 16:30:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-25 07:07:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-03-26 11:31:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c1913f1a1a52a7d93b9487c08711f75f4df6c697","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-24 15:58:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4de13d4b89745e0578b6760f80bb5c7b60464cd9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-24 16:29:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98160cde99c714c66defdbb516c8a09f014bfad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 16:30:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29851bb6824e4ced91954915eea44523 : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d030c4d113745d1ba278ab547546120 : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85089a079a474320aa7a582f59235db3 : SUCCESS in 11m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce999e3c7e3f4078b03a5fdf51ad0809 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26cfbccd01345e98c33d3744d5c58dd : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/151fd7c852b7497aacdd0867787705c8 : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb3c48fc34cc41e18b25ef586dda842b : FAILURE in 6m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bdfda893e9e4efca154df927ae43612 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0bb96a3f4dbc8b0cc16e313e5e6369d56e80004e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-25 07:07:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2789f5b6be142ba41d1bffb166dfec26797698eb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-26 11:31:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79dd3dfa84cd9f17ad837d1fa373ae6b9cdbd5bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 11:46:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db78229c639cc5165bfdda1b2659d72d217b9dac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 12:08:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a0a677584d4a46a0bda9d57f7c5ed0 : SUCCESS in 9m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9869744c8f94452eaca6724c5760463c : SUCCESS in 12m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21c3a737c9724d8fa2462bb1f02b2adc : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8479219cee54f45b7a94db4e120a055 : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28051104047745bb93374319104afb46 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"944e2ddfcec70ff1fa6fde75c53e79faf7fbe1fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 12:08:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"289f74baaf1273c7cd5c6c3a2a2f76b4d011a639","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 12:12:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709ff9cac12745b99fa09057bb56ea5c : SUCCESS in 3m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e9376025daa4d8da5972b6245020ddb : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af1562eecb3bcbfa9ce83d785e15ee9c01b0494e","revisions":{"af1562eecb3bcbfa9ce83d785e15ee9c01b0494e":{"kind":"REWORK","_number":1,"created":"2021-03-24 15:58:44.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/06/782806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/06/782806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/06/782806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/06/782806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/06/782806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/06/782806/1"}}},"commit":{"parents":[{"commit":"e0d2f86bf85c5a8b90adb7f9df88e214eac0227a","subject":"Update .gitreview for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e0d2f86bf85c5a8b90adb7f9df88e214eac0227a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-24 15:58:41.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-24 15:58:41.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","message":"Update TOX_CONSTRAINTS_FILE for stable/wallaby\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/wallaby branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ibdf843347ab05abccdd3d2745d378c603313ca18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/af1562eecb3bcbfa9ce83d785e15ee9c01b0494e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/af1562eecb3bcbfa9ce83d785e15ee9c01b0494e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
