)]}'
{"id":"openstack%2Fnetworking-bagpipe~786074","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fwallaby~I1faf499f58cae55f26f09d994c592c0aac95291f","project":"openstack/networking-bagpipe","branch":"stable/wallaby","topic":"neutron-wallaby","hashtags":[],"change_id":"I1faf499f58cae55f26f09d994c592c0aac95291f","subject":"Switch tempest jobs for stable/wallaby branch to wallaby jobs","status":"MERGED","created":"2021-04-13 16:03:45.000000000","updated":"2021-04-19 16:03:20.000000000","submitted":"2021-04-19 15:59:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786074-neutron-wallaby","meta_rev_id":"995c35df321d2f8cb49d099c68645434f3998cbb","_number":786074,"virtual_id_number":786074,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-19 15:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-04-15 18:33:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-04-19 14:10:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-04-19 14:10:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-13 17:32:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 06:07:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-04-14 06:07:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-04-14 06:07:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"797c73458546eb90106400158841d75d214114b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-13 16:03:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6f90258656b7755db313bf87c9cfa140629156e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 17:32:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102ded73ebac46c1a94e1ebf1f684fc6 : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a50ff944e6b64be68d1f0f0b0ab32274 : SUCCESS in 9m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a10cad684a4a88919bf48f88b60914 : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0289fc15c4b44431a18d144aa068be8c : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfeebae0fbc8428287789a287a415016 : SUCCESS in 9m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a481950402441c1817634820bbb7b24 : FAILURE in 5m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09ec624d21f04a63aff15a43a0e95a62 : SUCCESS in 10m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/dff9ed4192944df384b73e4243ad011f : SUCCESS in 1h 11m 10s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/4bd4345eb12e41ceba496bd141ad14d6 : FAILURE in 16m 43s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/8a414b4430b245bdb514df74b774f541 : SUCCESS in 1h 26m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3082d86418f3b93e3cff501fc28facd9edd0563a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-15 18:33:28.000000000","message":"Patch Set 1: Code-Review+2\n\nThe patch linked in depends-on has merged. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"389694ff59a29e7384a260568402c036a076acdc","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-04-19 14:10:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06962e9342e8555b1b455b7f49b9ab06cd5ef8b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 14:10:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e85f3c64ed32a3139c1852eb3231739183f6e8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 15:59:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea2035d65b749b092085f943662e9a5 : SUCCESS in 9m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/418ae34ec2524f6480c8829d00da0587 : SUCCESS in 9m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac2b2faeca044939a05ca7e7286ce50d : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54540e7b104b4b3fa4a5f2dc3f611782 : SUCCESS in 8m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/62a4d5394a2d42df8ac0c57c67056539 : SUCCESS in 1h 18m 30s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/d9a43d1f0fe440baad8d08a9a3a50007 : SUCCESS in 1h 41m 17s","accounts_in_message":[],"_revision_number":1},{"id":"47d0b1de6033d636fe2d1151c885106afa072ff4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 15:59:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"995c35df321d2f8cb49d099c68645434f3998cbb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 16:03:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92cd5583068b4e38affa439f28b1d952 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"899135705e9bd8585a2a203a1aed608309215d8d","revisions":{"899135705e9bd8585a2a203a1aed608309215d8d":{"kind":"REWORK","_number":1,"created":"2021-04-13 16:03:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/74/786074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/74/786074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/786074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/786074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/786074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/74/786074/1"}}},"commit":{"parents":[{"commit":"af1562eecb3bcbfa9ce83d785e15ee9c01b0494e","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/af1562eecb3bcbfa9ce83d785e15ee9c01b0494e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-04-13 16:03:05.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-04-13 16:03:05.000000000","tz":120},"subject":"Switch tempest jobs for stable/wallaby branch to wallaby jobs","message":"Switch tempest jobs for stable/wallaby branch to wallaby jobs\n\nChange-Id: I1faf499f58cae55f26f09d994c592c0aac95291f\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/785092\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/899135705e9bd8585a2a203a1aed608309215d8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/899135705e9bd8585a2a203a1aed608309215d8d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
