)]}'
{"id":"openstack%2Fnetworking-bagpipe~791126","triplet_id":"openstack%2Fnetworking-bagpipe~master~I3b8cbc5a8a8abce6ec2432aa78921656d20efe8a","project":"openstack/networking-bagpipe","branch":"master","topic":"ovn-devstack","hashtags":[],"change_id":"I3b8cbc5a8a8abce6ec2432aa78921656d20efe8a","subject":"Explicitly set job networking-bagpipe-tempest to ML2/OVS","status":"MERGED","created":"2021-05-13 09:38:23.000000000","updated":"2021-06-01 17:03:00.000000000","submitted":"2021-06-01 16:53:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"791126-ovn-devstack","meta_rev_id":"18b98b9f81ccc4f321ba7e2a8e378595321ab98a","_number":791126,"virtual_id_number":791126,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 16:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-06-01 15:26:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-06-01 15:26:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-06-01 15:26:07.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-06-01 15:26:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 10:06:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 14:33:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-19 07:49:27.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-05-25 15:05:38.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"5dba741eafff293b319d1b4bf9281d7fd2405a4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-13 09:38:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84e254a0928cd0c19fcef2af459ce2724d391898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 10:06:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a6d602101a04f34a2901b6271d50949 : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/798679aa2cc14c688a51ed9e658a439f : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b05a463418014834b6436a3224947ece : SUCCESS in 9m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a234961f1882437983c668353e90bb14 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd50602a41c47b3b8fe821fa5787e1b : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46c7f56dadf84130a9d9c7ffab533ab3 : FAILURE in 4m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c7e566bb0fb4189bbce7153a158ea7f : SUCCESS in 9m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/edf689e8c86446588f9e8cab1e965e66 : SUCCESS in 23m 44s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/50ac1c4d35994d548b7b361e8abd177f : FAILURE in 20m 02s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/441f4b50fa4a48bd9f908a9eac75f681 : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e2e3570aaf636b56a32d7e39c621c2e25c55d11c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-14 11:32:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2fc4cb8affd5f5c50d34da90a2e871587b720c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 12:50:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34951e6d1ffe4228aa73cddbfca9d651 : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c97ee918234474ea100bc383a370a90 : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/331b6b7d1bde45f19a3772604b611ff0 : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33d443e068514894b75ced7b77e62193 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d1c0be407c46edb66a4122ea059cb3 : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/987114ef020e4ca7856c2adf4504e84a : FAILURE in 4m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1749fe8f26344d2abb4b7a68e8db2d1a : SUCCESS in 10m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f6dea966dcc14d3f9c2b72b18dd64c7f : SUCCESS in 56m 44s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/da74a13386f64969a7c38649e36c6c6a : FAILURE in 16m 43s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/dbd3a191891546f3bc0e6a72a33e5dce : SUCCESS in 1h 13m 39s","accounts_in_message":[],"_revision_number":2},{"id":"addcf8099c080837a6fc0ab1a7025f8b165ab3e1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-14 14:33:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"11ef2aa1bece58ce59912fe81864fbb252e9962a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-19 07:49:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8a415e20a1e09997bcd2b5aff982e21611e474cf","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-25 15:05:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a126e1ea17e34958171650322789c77615a03b90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 15:06:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5bddb235777bc58d3a37f28ca417272c6b6192dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:05:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62733b4712ba47ca918b66223ed56e0c : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f7a4012aec744348c335456e5c321a6 : SUCCESS in 10m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8593bba8ba24bf087cfe11a88a6ad90 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ca35984acf43d1b1dce1a049e97087 : SUCCESS in 8m 21s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1890c6c2fbef4d7fae84a47a4d4cd717 : SUCCESS in 58m 15s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/5865a4d65b0a4cf58140029e4e27d5b9 : FAILURE in 55m 43s","accounts_in_message":[],"_revision_number":2},{"id":"135c07ae4f5cfe4f6a98da8b4e3502e6d7cc3de4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-26 04:39:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6a8e508f2ffb3a210067fed7465da6033c935aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 05:50:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4381b6bda745cd9a01c98b810b9250 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cffb6d1227f4bb3a855f65a0e9e9ab4 : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f577983ef374e6eaec33c101c104e15 : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41b13fd80307453aade5adb50b350237 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e1691423c7421abbd77a87eb492833 : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e98a8b03864eb39a87c17b512e2057 : FAILURE in 5m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c2bfbc9289c4474b86933cb07fe13c3 : SUCCESS in 9m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e34fd10de8c94384810d9f5f1c1f7a2c : SUCCESS in 1h 08m 51s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/4ecbc118585a4e9fb854a86537df89dc : FAILURE in 15m 53s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/77244b8cfd7047eaa79fa50352871781 : FAILURE in 55m 10s","accounts_in_message":[],"_revision_number":2},{"id":"11ed8510ca345c72f660d6643e40f74d7fc14d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-26 06:15:02.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"359f7a27b3a3dfb61035ccae512bcf7175c9f61e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-26 06:16:23.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"9bdd4f77270f5e462a7c2079f0315c563f64ac1b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-26 06:17:39.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8d2a4fe9907086b3eb882fd07663e086a8eb0ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 07:38:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e63040cb1cda434fb21090d03889a716 : SUCCESS in 11m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b37633f9e6b4a2e9f10c0ff29280002 : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16ebc2ad18804ae8b54dfc1293a4c02f : SUCCESS in 10m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61258560df29442bbf0b4de560782a01 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be282841d7f94ce69f6028bd8300e30b : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8c93edcc75c4ed4ab33a59776c23155 : FAILURE in 7m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3db1d44b7f54ffd9e07059b1d145dfb : SUCCESS in 12m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/11f77286abc74267a14c15c9175fdda2 : SUCCESS in 1h 15m 00s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a636e885156b45cf826bfeadcb461066 : FAILURE in 22m 29s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/196b476cdeda4495bf863d25dbca5b64 : RETRY_LIMIT in 11m 24s","accounts_in_message":[],"_revision_number":4},{"id":"fce488f5b158a684de33c17dd2a6bac813a6caa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-28 06:56:59.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0ae261eb59e37f514222906a23bdd24a4fed036d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 08:00:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f8d8a5940fe4a229eb929df83fe6dfc : SUCCESS in 9m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4510708f1b74888b21bb2a0ceb36f11 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6c547b8bed44a9b90e8d8a83f3629b1 : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27d106e6c38943218456fd0ee386a1db : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83851db272334e87afe9e63051cdd4ee : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c869657a3d64c3b82009d612c213895 : FAILURE in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ca29295328f4d92aaca6713739cdc3b : SUCCESS in 10m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8e02bbcc76c94513ac5f75656831ad88 : SUCCESS in 1h 00m 50s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/cc2b91cb1f424d02bdb5204a84f2f425 : FAILURE in 15m 26s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/be86fa8df10443829f530a30dc8b329b : FAILURE in 46m 26s","accounts_in_message":[],"_revision_number":5},{"id":"5e13b5c187ef8a0605c193d57e201b95a58db470","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-28 08:06:19.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6f559a2c1a45a05f07ecd1c6e3b19fb11ef42c9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-31 09:00:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ccb96b167f09a93a873f981381a5755fe2959a97","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-31 09:00:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e02e41108ac6757606ed9704c60b73265a383bb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 10:31:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1714d9e3d84f4ba0a36b03b432c00d : SUCCESS in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d1253bad79742a4979cbbeb1f4ce9ec : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0aa23dbd3a2a4d44b799408a8e7751a3 : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/955fd5ee22204b198208e8c887bba7a1 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0647586dc7e94d64aa41a1f6e11665fd : SUCCESS in 8m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/994a0d8033bb4d4dabba714f26b432db : FAILURE in 4m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0617c0f8b00b42e9bac67b8ffb58286a : SUCCESS in 9m 31s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d3cc31f83fd34d2fb03d9d0ac423ef6a : SUCCESS in 1h 12m 02s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/4f3423f3d94a4b8f8b418792148bd5e5 : FAILURE in 15m 13s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ea465cad4201471194834d3c15bc8ee7 : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":6},{"id":"500e340f3824be0682247626ba55500496f7cf8d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-01 15:26:07.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"23b845aead488f6d1f370c1f706ee543f8ad72ec","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-01 15:26:32.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"358fc97b952dbc2e055affebf9471f8cd4261381","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-01 15:26:51.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"00ab9441ebfc31a5849810c8983e1ca1105ed40a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 15:32:26.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"663ac4380d46268e91c2d49cecdf973e201fd858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:53:58.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b582711c5e54e2cb91b64ee3965ac7c : SUCCESS in 9m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac5abb9b04ec4dedb7d424608336e009 : SUCCESS in 9m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4dbd2e78da0483985a1a52572bbfb15 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a7c67e1cb341848abd4a4c05b3df69 : SUCCESS in 8m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5fba6b9b98564a0ab1a0025412fb9fda : SUCCESS in 1h 14m 26s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1c42b4e4a50d418c8dcc15a302e44e0e : SUCCESS in 1h 08m 18s","accounts_in_message":[],"_revision_number":6},{"id":"40c4234acad5d8598342a5f5b3df04affb7c242a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:53:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"18b98b9f81ccc4f321ba7e2a8e378595321ab98a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 17:03:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d21a2be6494575ad426f7fc098ca14 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d7506857dea5dba51ba5ef4e4694b9e16ef1f20e","revisions":{"48727ddaa552d82185aa764f9ede1076cc9e9d15":{"kind":"REWORK","_number":1,"created":"2021-05-13 09:38:23.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/26/791126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/791126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/1"}}},"commit":{"parents":[{"commit":"c3e941f99405d06d943421374e19ff63e4d4a57c","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c3e941f99405d06d943421374e19ff63e4d4a57c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-05-13 09:35:55.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-05-13 09:35:55.000000000","tz":120},"subject":"Explicitly set job networking-bagpipe-tempest to ML2/OVS","message":"Explicitly set job networking-bagpipe-tempest to ML2/OVS\n\nDevstack is changing the Neutron default to OVN backend.\nThis patch makes sure that networking-bagpipe-tempest job use OVS\nbackend.\n\nChange-Id: I3b8cbc5a8a8abce6ec2432aa78921656d20efe8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/48727ddaa552d82185aa764f9ede1076cc9e9d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/48727ddaa552d82185aa764f9ede1076cc9e9d15"}]},"branch":"refs/heads/master"},"0d29271c360c9926c3f87e1c7ced106d8533cd79":{"kind":"REWORK","_number":2,"created":"2021-05-14 11:32:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/26/791126/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/791126/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/2"}}},"commit":{"parents":[{"commit":"c3e941f99405d06d943421374e19ff63e4d4a57c","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c3e941f99405d06d943421374e19ff63e4d4a57c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-05-13 09:35:55.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-05-14 11:32:28.000000000","tz":0},"subject":"Explicitly set job networking-bagpipe-tempest to ML2/OVS","message":"Explicitly set job networking-bagpipe-tempest to ML2/OVS\n\nDevstack is changing the Neutron default to OVN backend.\nThis patch makes sure that networking-bagpipe-tempest job use OVS\nbackend.\n\nChange-Id: I3b8cbc5a8a8abce6ec2432aa78921656d20efe8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0d29271c360c9926c3f87e1c7ced106d8533cd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0d29271c360c9926c3f87e1c7ced106d8533cd79"}]},"branch":"refs/heads/master"},"8b6869d5131e22678c666840896a424c7c434feb":{"kind":"REWORK","_number":3,"created":"2021-05-26 06:15:02.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/26/791126/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/791126/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/3"}}},"commit":{"parents":[{"commit":"c3e941f99405d06d943421374e19ff63e4d4a57c","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c3e941f99405d06d943421374e19ff63e4d4a57c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-05-13 09:35:55.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-05-26 06:14:56.000000000","tz":0},"subject":"Explicitly set job networking-bagpipe-tempest to ML2/OVS","message":"Explicitly set job networking-bagpipe-tempest to ML2/OVS\n\nDevstack is changing the Neutron default to OVN backend.\nThis patch makes sure that networking-bagpipe-tempest job use OVS\nbackend.\n\nChange-Id: I3b8cbc5a8a8abce6ec2432aa78921656d20efe8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8b6869d5131e22678c666840896a424c7c434feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8b6869d5131e22678c666840896a424c7c434feb"}]},"branch":"refs/heads/master"},"7850869d9dda1d5311c931fa1c593d07f9b856b8":{"kind":"REWORK","_number":4,"created":"2021-05-26 06:16:23.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/26/791126/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/791126/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/4"}}},"commit":{"parents":[{"commit":"c3e941f99405d06d943421374e19ff63e4d4a57c","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c3e941f99405d06d943421374e19ff63e4d4a57c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-05-13 09:35:55.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-05-26 06:16:20.000000000","tz":0},"subject":"Explicitly set job networking-bagpipe-tempest to ML2/OVS","message":"Explicitly set job networking-bagpipe-tempest to ML2/OVS\n\nDevstack is changing the Neutron default to OVN backend.\nThis patch makes sure that networking-bagpipe-tempest job use OVS\nbackend.\n\nChange-Id: I3b8cbc5a8a8abce6ec2432aa78921656d20efe8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7850869d9dda1d5311c931fa1c593d07f9b856b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7850869d9dda1d5311c931fa1c593d07f9b856b8"}]},"branch":"refs/heads/master"},"c7f29fc57e0abe149bfcb4687fd37ca1a834c707":{"kind":"REWORK","_number":5,"created":"2021-05-28 06:56:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/26/791126/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/791126/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/5"}}},"commit":{"parents":[{"commit":"c3e941f99405d06d943421374e19ff63e4d4a57c","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c3e941f99405d06d943421374e19ff63e4d4a57c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-05-13 09:35:55.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-05-28 06:56:54.000000000","tz":0},"subject":"Explicitly set job networking-bagpipe-tempest to ML2/OVS","message":"Explicitly set job networking-bagpipe-tempest to ML2/OVS\n\nDevstack is changing the Neutron default to OVN backend.\nThis patch makes sure that networking-bagpipe-tempest job use OVS\nbackend.\n\nChange-Id: I3b8cbc5a8a8abce6ec2432aa78921656d20efe8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c7f29fc57e0abe149bfcb4687fd37ca1a834c707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c7f29fc57e0abe149bfcb4687fd37ca1a834c707"}]},"branch":"refs/heads/master"},"d7506857dea5dba51ba5ef4e4694b9e16ef1f20e":{"kind":"REWORK","_number":6,"created":"2021-05-31 09:00:14.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/26/791126/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/791126/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/791126/6"}}},"commit":{"parents":[{"commit":"c3e941f99405d06d943421374e19ff63e4d4a57c","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c3e941f99405d06d943421374e19ff63e4d4a57c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-05-13 09:35:55.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-05-31 08:59:54.000000000","tz":120},"subject":"Explicitly set job networking-bagpipe-tempest to ML2/OVS","message":"Explicitly set job networking-bagpipe-tempest to ML2/OVS\n\nDevstack is changing the Neutron default to OVN backend.\nThis patch makes sure that networking-bagpipe-tempest job use OVS\nbackend.\n\nChange-Id: I3b8cbc5a8a8abce6ec2432aa78921656d20efe8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d7506857dea5dba51ba5ef4e4694b9e16ef1f20e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d7506857dea5dba51ba5ef4e4694b9e16ef1f20e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
