)]}'
{"id":"openstack%2Fnetworking-bagpipe~796477","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Ftrain~Icec5d9be8d78408ad1c758573041225777e86d48","project":"openstack/networking-bagpipe","branch":"stable/train","topic":"clean-non-voting-jobs","hashtags":[],"change_id":"Icec5d9be8d78408ad1c758573041225777e86d48","subject":"[EM releases] Move non-voting jobs to the experimental queue","status":"MERGED","created":"2021-06-15 13:51:31.000000000","updated":"2021-08-17 16:48:26.000000000","submitted":"2021-08-17 16:44:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"796477-clean-non-voting-jobs","meta_rev_id":"a5eff7396d48cc986dcf2700ebc2361d21d1e1b1","_number":796477,"virtual_id_number":796477,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-17 16:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-08-17 15:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-07-01 08:47:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-06-23 08:00:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2021-07-06 12:42:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-17 15:11:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-08-17 15:12:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 14:15:47.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-06-15 14:16:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-06-15 15:26:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 14:05:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-06-22 09:29:22.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-06-23 08:00:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-07-06 09:01:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-07-06 09:01:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-17 15:11:57.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"466136e59928f8fa2f931db15e66393fcb14fc01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-15 13:51:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ab3a2ce66fa83e6daad7475a7c33a3208b7d8d3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-15 14:15:47.000000000","message":"Patch Set 1:\n\nLGTM, let\u0027s see how CI is doing on voting jobs","accounts_in_message":[],"_revision_number":1},{"id":"6882e426434fe843475b45b18b1ee9161b401180","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-15 14:16:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c46c21c064041c7740651bff1e2aaeb41b1547c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:26:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5daf8700ef464979aed64003136799c9 : SUCCESS in 9m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5e050a9ff024167a066d915bad5a62c : FAILURE in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3994d8fd8c1b45b2b58e07301b9c1cc9 : SUCCESS in 9m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d663b8ab6ed04942aee36f466fb4d53a : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e005ce70a4d4525ad2b926eaec42369 : FAILURE in 6m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df1772723a9e4819b59daf9e009eaea2 : SUCCESS in 8m 26s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/fc8b587b23db409584a9f9e50fe067fc : SUCCESS in 1h 15m 51s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6969437cd1584d6e9919c028d181a406 : SUCCESS in 1h 28m 22s","accounts_in_message":[],"_revision_number":1},{"id":"86ab8505dcb85c6c023c2c242e2008a514b47422","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-16 14:05:02.000000000","message":"Patch Set 1:\n\nIs it possible to constrain Neutron to be \u003c16.0.0 ? Doc build fails as it installs neutron\u003d18.0.0 and can\u0027t import neutron.objects.common_types, which is there if neutron is less than 16.0.0","accounts_in_message":[],"_revision_number":1},{"id":"399d8fb86e93d7d46de79c1944851855079e704d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-16 14:22:23.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Is it possible to constrain Neutron to be \u003c16.0.0 ? Doc build fails as it installs neutron\u003d18.0.0 and can\u0027t import neutron.objects.common_types, which is there if neutron is less than 16.0.0\n\nIt should, docs tox target has global upper-constraints, and I see the requirements repo cloned and checked out to stable/train, so it should be neutron\u003d\u003d\u003d15.3.4 ?","accounts_in_message":[],"_revision_number":1},{"id":"4533ada26f114e2a82fec0fc93af551fb511fbf9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-16 14:23:50.000000000","message":"Patch Set 1:\n\nAnd apparently someone I will not name had a fix for that few months ago:\nhttps://opendev.org/openstack/networking-bagpipe/commit/f58058aa06eb5648950e9c2a6f663843570ed8cd","accounts_in_message":[],"_revision_number":1},{"id":"41c26cadcb2b47e0ed0a9bc685fe0cfb8b55e3a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-17 08:09:16.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"626bf64b0904aaca41f406979371163b2bd31098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 09:30:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156245dd785446b0bd9419834f05557e : SUCCESS in 9m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/337e5e88bc9e429a95a58825836ad287 : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6da5d96b9d734f948c0ce4af99c5ae21 : SUCCESS in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b8c18b812f4450dbf81288166dca839 : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e3a7aa0e234838a50481b0467a8535 : FAILURE in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2793b8aa5a3484eae96fd4fc52f8adb : SUCCESS in 5m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/083c5f0abc9148369089f14234e3f156 : RETRY_LIMIT in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23b6cde48c264edcbbff9327445f2ae1 : SUCCESS in 8m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/d0a23af66d2d4366acda5edd3457c06e : SUCCESS in 1h 08m 03s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/19dfeae4dd71490baf0a2c1af2902d4e : SUCCESS in 1h 16m 15s","accounts_in_message":[],"_revision_number":2},{"id":"1d0f16ea69774dfc031633fe48f63fd704eca9c4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-18 09:26:24.000000000","message":"Patch Set 2:\n\nStrange, even with upper-req ad that change to doc/requirements.txt I still see neutron\u003d\u003d18.0.0 in log.","accounts_in_message":[],"_revision_number":2},{"id":"f4846836d4e34ce049ee5af86b6f0ee4b79c983c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-22 09:29:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d1bb0ad53acf014433c9b54302f906a96f0b190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-22 09:31:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fba0063274ebc75f495dc5518f3a80b6e51e84d8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-22 09:50:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"78d823c60a8852333fc82a6864d8f0ac475e0200","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 11:12:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd61d78ce3224a65b4fd988895280df8 : SUCCESS in 9m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d72fcd54ae024c2bbca87ff96d424a09 : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dc4d656b4e84ce7bef7ec068c6cf719 : SUCCESS in 8m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bede4ef054443f987f5a4ca1dc03473 : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4728d44341c43d79050850a0eff203f : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a30bbf94947421cb3e894be950458c0 : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be90e67e2c104c3cac7dc8a8b1572c04 : SUCCESS in 8m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/63a6203aa9e84e988894e8ddcaffe947 : SUCCESS in 1h 12m 45s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/81bf1005c73e41a7aa0fc1f70a0e31d4 : SUCCESS in 1h 34m 52s","accounts_in_message":[],"_revision_number":4},{"id":"20ab913afa27a62c616d157e6d439905b88d01a3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-23 08:00:58.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a05ac09a28d774bd2b0b98697f6becc88e1ed892","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-07-01 08:47:35.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks for explanation and fix Elod!","accounts_in_message":[],"_revision_number":4},{"id":"242fc1e920c54c14d88aa6d3172dee875ecfac5e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-06 12:42:03.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3130cabb1483b3e64928afebbb563f51bfb97e12","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-08-17 15:11:57.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"accfd2f51230dcb37c02c947d08d4ab86ed84e4d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-17 15:12:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"364e0ad60faf4eae44103e5e6b6e8af010cc6cde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 15:13:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8e057b3cc145201c7a40eb3f561c89f85fc6018e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:44:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e3bceafd23447b9f4118cba7a09f6f : SUCCESS in 8m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9fb25c19590448c9ab062916d38ec492 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8426355df63e4890a7ce1ae1df732983 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d58a1a7b9c864e748437ca2608f52c18 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d84d1f6d40a490baed63664653f18e4 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44bcb520ee174c2bbbc0d17a8539e5e7 : SUCCESS in 4m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/eeb8e58e13df48d4bbe4afe1fdf8805a : SUCCESS in 1h 13m 55s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/2579d0c12daa46ac8a8c6ab69b7e0c98 : SUCCESS in 1h 28m 03s","accounts_in_message":[],"_revision_number":4},{"id":"52697787ca6be75ff0408d9cd6514886fbd8a377","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:44:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a5eff7396d48cc986dcf2700ebc2361d21d1e1b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:48:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5e6eb2028742fabe1afc73dd03cf32 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56a6be528c4f4b60b3274dbd442b3321 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"65c389168b3414975844b3461f53a75aa4817c38","revisions":{"abbed1cda8e38859a1ebedc2ebd119b684593f20":{"kind":"REWORK","_number":1,"created":"2021-06-15 13:51:31.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/77/796477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/77/796477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/1"}}},"commit":{"parents":[{"commit":"2512a2b7a0553856d4bfd73d1b25542dd6939e91","subject":"Monkey patch original current_thread _active","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2512a2b7a0553856d4bfd73d1b25542dd6939e91"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-15 13:50:52.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-15 13:51:25.000000000","tz":120},"subject":"[EM releases] Move non-voting jobs to the experimental queue","message":"[EM releases] Move non-voting jobs to the experimental queue\n\nDuring the CI meeting we agreed that non-voting jobs in the branches\nwhich are in Extened Maintenance (EM) phase should be moved to from\nthe check to the experimental queue.\n\nThis patch is doing exactly that.\n\nChange-Id: Icec5d9be8d78408ad1c758573041225777e86d48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/abbed1cda8e38859a1ebedc2ebd119b684593f20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/abbed1cda8e38859a1ebedc2ebd119b684593f20"}]},"branch":"refs/heads/stable/train"},"f66d6cf79b2ac24c3f4e384189d42e750acb89dd":{"kind":"REWORK","_number":2,"created":"2021-06-17 08:09:16.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/77/796477/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/77/796477/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/2"}}},"commit":{"parents":[{"commit":"2512a2b7a0553856d4bfd73d1b25542dd6939e91","subject":"Monkey patch original current_thread _active","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2512a2b7a0553856d4bfd73d1b25542dd6939e91"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-15 13:50:52.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-06-17 08:09:08.000000000","tz":0},"subject":"[EM releases] Move non-voting jobs to the experimental queue","message":"[EM releases] Move non-voting jobs to the experimental queue\n\nDuring the CI meeting we agreed that non-voting jobs in the branches\nwhich are in Extened Maintenance (EM) phase should be moved to from\nthe check to the experimental queue.\n\nThis patch is doing exactly that.\n\nChange-Id: Icec5d9be8d78408ad1c758573041225777e86d48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f66d6cf79b2ac24c3f4e384189d42e750acb89dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f66d6cf79b2ac24c3f4e384189d42e750acb89dd"}]},"branch":"refs/heads/stable/train"},"323d5c99fb99050c4147c8e4564e9404bcbd1bca":{"kind":"REWORK","_number":3,"created":"2021-06-22 09:29:22.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/77/796477/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/77/796477/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/3"}}},"commit":{"parents":[{"commit":"2512a2b7a0553856d4bfd73d1b25542dd6939e91","subject":"Monkey patch original current_thread _active","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2512a2b7a0553856d4bfd73d1b25542dd6939e91"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-15 13:50:52.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-22 09:28:19.000000000","tz":120},"subject":"[EM releases] Move non-voting jobs to the experimental queue","message":"[EM releases] Move non-voting jobs to the experimental queue\n\nDuring the CI meeting we agreed that non-voting jobs in the branches\nwhich are in Extened Maintenance (EM) phase should be moved to from\nthe check to the experimental queue.\n\nThis patch is doing exactly that.\n\nChange-Id: Icec5d9be8d78408ad1c758573041225777e86d48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/323d5c99fb99050c4147c8e4564e9404bcbd1bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/323d5c99fb99050c4147c8e4564e9404bcbd1bca"}]},"branch":"refs/heads/stable/train"},"65c389168b3414975844b3461f53a75aa4817c38":{"kind":"REWORK","_number":4,"created":"2021-06-22 09:31:01.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/77/796477/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/77/796477/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/77/796477/4"}}},"commit":{"parents":[{"commit":"2512a2b7a0553856d4bfd73d1b25542dd6939e91","subject":"Monkey patch original current_thread _active","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2512a2b7a0553856d4bfd73d1b25542dd6939e91"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-15 13:50:52.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2021-06-22 09:30:39.000000000","tz":120},"subject":"[EM releases] Move non-voting jobs to the experimental queue","message":"[EM releases] Move non-voting jobs to the experimental queue\n\nDuring the CI meeting we agreed that non-voting jobs in the branches\nwhich are in Extened Maintenance (EM) phase should be moved to from\nthe check to the experimental queue.\n\nThis patch is doing exactly that.\n\nChange-Id: Icec5d9be8d78408ad1c758573041225777e86d48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/65c389168b3414975844b3461f53a75aa4817c38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/65c389168b3414975844b3461f53a75aa4817c38"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
