)]}'
{"id":"openstack%2Fnetworking-bagpipe~823255","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ia4d56b34967cb44ff6483e5c5c2a0587c8f44f63","project":"openstack/networking-bagpipe","branch":"master","attention_set":{},"removed_from_attention_set":{"32326":{"account":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"last_update":"2022-03-09 14:48:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia4d56b34967cb44ff6483e5c5c2a0587c8f44f63","subject":"Update python testing classifier","status":"MERGED","created":"2022-01-03 09:00:47.000000000","updated":"2022-03-09 14:49:47.000000000","submitted":"2022-03-09 14:48:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823255","meta_rev_id":"baeaaad40e75b9b3194d3324d69142c6b4510691","_number":823255,"virtual_id_number":823255,"owner":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-09 14:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-03-09 12:54:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-01-27 16:50:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-03-09 12:54:28.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-03-09 12:54:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2022-03-09 12:54:28.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-03 10:25:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 16:50:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-03-09 12:54:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-03-09 12:54:28.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"86e53a2efa292d8f1fb5c616e93a4f7b303a47d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"date":"2022-01-03 09:00:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68415f201ac22c182e6dd837ac0ea81c76b70b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 10:25:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fc149e7e357d49699e98b80dd47c84c0 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/632e85faa5f94213855c257e71fb2c6d : SUCCESS in 10m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b661cc97b7f04006823a74924aa2ff44 : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db93535952234d229fa64b4dde16b200 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/296b492c0d774c0c9bbf25c2c6c487c2 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29d10b7815274d978a1747119303526a : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea7c5e3fab7e425dac4133f188f7ce0e : FAILURE in 4m 13s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97ac81e0f66a407ab1542e0b1baaaf6d : SUCCESS in 8m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/906703f45e344a9da07ef3132fab75f5 : SUCCESS in 10m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e6dc3fd051ee466d827836af301d4746 : SUCCESS in 1h 06m 55s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e7a6c4aa373848ffb518b924f3f44826 : FAILURE in 14m 24s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/123401f76ac64b2f92ef6712a21382b0 : SUCCESS in 1h 22m 10s","accounts_in_message":[],"_revision_number":1},{"id":"79575c9f8b0ee571755ca01c1dce8252da8ca4df","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-27 16:50:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"88eea298b428c12785a605bd55df0fefeb37b542","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-09 12:54:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb97f5cfeb27e9dc037197f7afb9a85a1dbc4141","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-09 12:54:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"865bece1b545f5b19cddf0c1608c466b92876652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 12:55:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e752cc0c8146de7cba9d87ef61a8cb97118acfbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 14:48:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0031e15633894a2b8fed814b991f5fc5 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6920a97f30ab4f7fb90f722bf65d352c : SUCCESS in 11m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d134378770014b3ea42ae4ed89d5ee00 : SUCCESS in 10m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8bd0690d24246b1b62c2dc510aaf772 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd97dc0a08f465c92af5f816b5ee86e : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86ed2c01cac34e9a8ff941a0faa56b67 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/add4a08e3d774b37a3a8d86775aadda3 : SUCCESS in 9m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/649f7b0e81ec4a278d85a115246802d6 : SUCCESS in 1h 14m 58s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1ca264bd31cb48c99b07827a2c05a4db : SUCCESS in 1h 47m 56s","accounts_in_message":[],"_revision_number":1},{"id":"dddd2504957648cad69b19c729cef6b3aecffac8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 14:48:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"baeaaad40e75b9b3194d3324d69142c6b4510691","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 14:49:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d5aff7a9e149f6ac23bafc466ab41e : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9462498681c048c78bc2f1359e993e2a : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b56ef6305517d5d9f66e14fc6e264c8172af2cf0","revisions":{"b56ef6305517d5d9f66e14fc6e264c8172af2cf0":{"kind":"REWORK","_number":1,"created":"2022-01-03 09:00:47.000000000","uploader":{"_account_id":32326,"name":"YuehuiLei","email":"leiyuehui@inspur.com","username":"YuehuiLei"},"ref":"refs/changes/55/823255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/55/823255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/823255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/823255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/823255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/55/823255/1"}}},"commit":{"parents":[{"commit":"8a8279a61fbbf6ae70c8e7b25c89a6a6d71e898d","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8a8279a61fbbf6ae70c8e7b25c89a6a6d71e898d"}]}],"author":{"name":"YuehuiLei","email":"leiyuehui@inspur.com","date":"2022-01-03 08:58:15.000000000","tz":480},"committer":{"name":"YuehuiLei","email":"leiyuehui@inspur.com","date":"2022-01-03 08:58:23.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nYoga testing runtime[1] has been updated to add py39\ntesting as voting. Unit tests update are handled by the\njob template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nthis commit updates the classifier in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\nChange-Id: I4c9fd157507bab63bfd6821d8476daa625992d72\n\nChange-Id: Ia4d56b34967cb44ff6483e5c5c2a0587c8f44f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b56ef6305517d5d9f66e14fc6e264c8172af2cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b56ef6305517d5d9f66e14fc6e264c8172af2cf0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
