)]}'
{"id":"openstack%2Fnetworking-bagpipe~825247","triplet_id":"openstack%2Fnetworking-bagpipe~master~I8d8846819d694fb9356f47bbbe4188cedde054c5","project":"openstack/networking-bagpipe","branch":"master","topic":"train-tempest-cap","hashtags":[],"change_id":"I8d8846819d694fb9356f47bbbe4188cedde054c5","subject":"DNM: test tempest 26.1.0 tag","status":"ABANDONED","created":"2022-01-19 03:56:00.000000000","updated":"2022-01-27 16:36:37.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0c4519f91ef42307a1913b5df555d0fcbbed7988","_number":825247,"virtual_id_number":825247,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-19 05:27:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 05:27:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29f758206d697fef0599b31c5e5bfc9b8f804902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-19 03:56:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dec078da60ba0028488505a3f2038b3f7d9e66e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 05:27:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c64b968269846c087956f3611070a29 : SUCCESS in 12m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0faba79c2130447ebdd4426b9850f78f : SUCCESS in 10m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/021646112f6644cbaaa87ab9656dff22 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8887d9ae44f242ce92238dc826ed7f18 : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36470ed67c344d60bc35480808db955f : FAILURE in 5m 03s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/957678ecf4804b0eb3ad3273bcb4afaa : SUCCESS in 8m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9972cb13abd442f78a3ec16348bf40d5 : SUCCESS in 11m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f378143cd3b1421dbd9688374002785c : SUCCESS in 1h 10m 55s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/9a88eef6ebe549f79b4583455727d2ab : FAILURE in 20m 06s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/9afea1f0bd624ee5a804657fc9959635 : SUCCESS in 1h 25m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7301d756ab587b52099c20e5fc32398ee4468677","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-19 19:23:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9d640137e1677428152a966bc6036ecb9c5aa7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 21:08:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90862099357b42489f22b7731363442c : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3638bc34445941f5b9e92ab47bf60edd : SUCCESS in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5413c499f2fa42c987143552e34be056 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96fef389092d45dda9022dc07be583d7 : SUCCESS in 8m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eae3d0d532584657be5cba48e8a1c13b : FAILURE in 5m 16s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfdc8a0068af480ea8997ddfa209a559 : SUCCESS in 8m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2ea63ccb0074fd08d8c0e92361d011c : SUCCESS in 9m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ad1e47190f3947a586e1a7bd8a51f0c3 : SUCCESS in 1h 07m 34s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/538e34f0aa874741aa0108320d78c7f5 : FAILURE in 18m 09s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f6d1c447aab9412baa42bb1204da2b91 : SUCCESS in 1h 20m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b0c978ae2a45435bafeb9faed0e0d7f26921cbb6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-21 03:24:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97cd79535b8e815b22ec2a37595e65682ac906c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 05:33:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71fedf3c06b140e2ab9d254b474dab9e : SUCCESS in 10m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb66dabc29974d87971e80ddb3476802 : SUCCESS in 10m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e346da39f9f4f78b6187857ae0ef9e0 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ed00175fb4435abaaed8d6644610eb : SUCCESS in 11m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83e129ce60a2461499d2c5e5b49e41cc : FAILURE in 5m 39s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0a6d9632c82488a9ff8dab29aa71690 : SUCCESS in 9m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4410e1d880944125aa8a5f479160eaa3 : SUCCESS in 11m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c42453eb0eeb497ea210a0a4e13b2216 : SUCCESS in 1h 02m 53s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/517833c56065458f8eb9418ce29ddeaa : FAILURE in 16m 48s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/5f4a5f6676504e068ef8b61b023bc101 : SUCCESS in 1h 43m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0c4519f91ef42307a1913b5df555d0fcbbed7988","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-27 16:36:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5ea6aff2cec8f8513e7aba37a71bf21b5f59b33","revisions":{"c5ea6aff2cec8f8513e7aba37a71bf21b5f59b33":{"kind":"REWORK","_number":1,"created":"2022-01-19 03:56:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/825247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/47/825247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/825247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/825247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/825247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/825247/1"}}},"commit":{"parents":[{"commit":"8a8279a61fbbf6ae70c8e7b25c89a6a6d71e898d","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8a8279a61fbbf6ae70c8e7b25c89a6a6d71e898d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-19 03:55:45.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-19 03:55:45.000000000","tz":-360},"subject":"DNM: test tempest 26.1.0 tag","message":"DNM: test tempest 26.1.0 tag\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/822339\nChange-Id: I8d8846819d694fb9356f47bbbe4188cedde054c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c5ea6aff2cec8f8513e7aba37a71bf21b5f59b33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c5ea6aff2cec8f8513e7aba37a71bf21b5f59b33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
