)]}'
{"id":"openstack%2Fnetworking-bagpipe~840098","triplet_id":"openstack%2Fnetworking-bagpipe~master~I07b3fc75cb1bc80cc40e102d96e63f1381996360","project":"openstack/networking-bagpipe","branch":"master","topic":"drop-l-c","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-05-02 19:04:35.000000000","reason":"Lajos Katona replied on the change"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-23 10:44:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I07b3fc75cb1bc80cc40e102d96e63f1381996360","subject":"Drop lower-constraints.txt and its testing","status":"MERGED","created":"2022-05-01 01:29:44.000000000","updated":"2022-05-23 10:47:20.000000000","submitted":"2022-05-23 10:44:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840098-drop-l-c","meta_rev_id":"2fcb7e60860d0157438ec5d89e9a0d15ffda046a","_number":840098,"virtual_id_number":840098,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-23 10:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-05-19 07:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-05-02 19:04:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-05-19 07:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-01 02:56:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-02 09:33:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2022-05-02 19:04:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-05-19 07:52:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"c58443920a9894250f1152d1097909a8e01335f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-01 01:29:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb707be149bfceaf806ba3290466bc22be904fbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 02:56:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b5c1b20799242f3abc2286afbf2d018 : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/114e85542ad94a569d8c21b2eb1a9874 : SUCCESS in 9m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1bf85816eb14fc3b18ec2afbf53f271 : FAILURE in 7m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4a0b12d6e224dbaa504ee4c616f69be : FAILURE in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570e960c23f14518bbd55cd70f363693 : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a234e2aa436479e8ce8e9c5a7649a9c : SUCCESS in 4m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa902f1298f44e7090301b63b98f7471 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c365a23035b418186ada608c6bc42c3 : FAILURE in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc37358d54f5456486fe1b22a5ebb8dd : SUCCESS in 9m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/acca877ad03f4e2ca4509611fb9d4101 : SUCCESS in 1h 06m 24s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/4c753f97725540fb84c6509c01241700 : FAILURE in 16m 35s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/e4a894bbcd2b41308f114b04b9ed562c : SUCCESS in 1h 21m 10s","accounts_in_message":[],"_revision_number":1},{"id":"493f78f3f71d8902d1842f42b12fa478be5241a2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-02 09:33:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"404be9d45fbfb40073c545ea84d87e5faa33a40c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-02 18:10:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec51460bc168fe1c0b9e0c45fb08c5846c623233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-02 18:10:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"577e5824010fbe50db01fcc34aa0843ae2b5b412","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-02 19:04:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"288bd587dc871555f5373089da0911f1e788495f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 19:31:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655de41e9d9543909fbb1d78cec11728 : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b00a8eca126949afb832137cee1f163f : SUCCESS in 9m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56f2ad2b42874d2b91b691bdf3b11f05 : SUCCESS in 9m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09eafd6bae014bfa89bbe4e56b9619b4 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d6cceb5626416984e2ddda348c35d8 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bb9f5a364a04f57947cd05e223d7cf3 : SUCCESS in 4m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0cdd7397d844607abdf6c0cccdabc9d : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d07f7d97213c4d668e006ab4c52a0250 : SUCCESS in 7m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9116a7a87e674af081e88e15a82c5b5a : SUCCESS in 9m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/03170b9798be4595a80be5136e9ebd6f : SUCCESS in 1h 06m 18s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/424370c96fa6407dbe35c5068b36d926 : FAILURE in 11m 55s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/81a8a66738294a229457e8a9e0bb99d4 : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":2},{"id":"a63ae7272f549428ce987fd8c6089bf1fd670e47","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-19 07:52:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"50956af60903d1f0f77c46f84a9169db3ad522d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 07:54:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f936a3cd2a669e407e2adac3f246bc19b3a4d781","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 09:03:54.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"d12e7f5d84c37bd825871ff6887284b488a8f097","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-23 07:34:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7c98905f5a71cd02b18b7871e3e7773aa813869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 09:02:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d95d6fddc4840118def41946110d184 : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30415407729447d5b97e3ee7520afa3e : SUCCESS in 11m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/574e43adb49a46e5acfdd6546327b054 : SUCCESS in 9m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f0bb44c63cd434e9eda7701511ee238 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb8254b9802474f85b67a73415b27b2 : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59ce1b562d6642b49d39272198d7754b : SUCCESS in 4m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b2c5c830e4544f48f848ebe22653aa3 : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80887dfc15fe43c888833c734aab2e0c : SUCCESS in 10m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/87c1f12cd9f74d3c97e6ca9876395cda : SUCCESS in 1h 13m 44s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/79de1c9184204e4a9a2845b73fe68f60 : FAILURE in 18m 24s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/8778cc4b48c74b738a799c4f7de578c9 : SUCCESS in 1h 22m 48s","accounts_in_message":[],"_revision_number":2},{"id":"e59bdac3a06f3f09c90d4b367243b056feafd07e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 09:03:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a7516cbbe1d9585935dd8c70fe873dc403eb6efa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 10:44:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7335b729b67f496891828127519b4438 : SUCCESS in 10m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c72b121358554afd94172ab6e25e613b : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef1b2c21917d4984965b6b15bec7cdc4 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdea9ca4d4b64c47988fa9935be58737 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d833a5ea85a40198ae7c77e892b6385 : SUCCESS in 4m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec38233a28bf497d8b2b0f7bd0b5df4f : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70003e1a8bbc44a4bee6a3825287ba84 : SUCCESS in 11m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/04bd3ba99aae4c4db98a1f53db17dc24 : SUCCESS in 1h 23m 47s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/353b6407036b45e78c7b76cb1fb4e6af : SUCCESS in 1h 35m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d947f522a351b31d02565c825a46cee5fc165da3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 10:44:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2fcb7e60860d0157438ec5d89e9a0d15ffda046a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 10:47:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64ef0fff7ed4d73beb4241cc9926c46 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f05ce303fac464cae3a3c6acb8c4ea5 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5f79987aec97d2d1dd5c26c8f8271f98a15bc60a","revisions":{"245d72bbc2bfd6a386836bc0d1f404e3bc8fccc0":{"kind":"REWORK","_number":1,"created":"2022-05-01 01:29:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/98/840098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/840098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/840098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/840098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/840098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/840098/1"}}},"commit":{"parents":[{"commit":"85d6c25c2f2f3a0434396c30b120189214dfc308","subject":"Add weekly jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/85d6c25c2f2f3a0434396c30b120189214dfc308"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 01:06:43.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 01:06:43.000000000","tz":-300},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I07b3fc75cb1bc80cc40e102d96e63f1381996360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/245d72bbc2bfd6a386836bc0d1f404e3bc8fccc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/245d72bbc2bfd6a386836bc0d1f404e3bc8fccc0"}]},"branch":"refs/heads/master"},"5f79987aec97d2d1dd5c26c8f8271f98a15bc60a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-02 18:10:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/98/840098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/98/840098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/840098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/840098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/840098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/98/840098/2"}}},"commit":{"parents":[{"commit":"c911fc41ac906e143dee836fcbf064726b436a6a","subject":"CI: add required projects for unit test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c911fc41ac906e143dee836fcbf064726b436a6a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 01:06:43.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-05-02 18:10:50.000000000","tz":0},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I07b3fc75cb1bc80cc40e102d96e63f1381996360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5f79987aec97d2d1dd5c26c8f8271f98a15bc60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5f79987aec97d2d1dd5c26c8f8271f98a15bc60a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
